)]}'
{"id":"openstack%2Fpuppet-gnocchi~782743","triplet_id":"openstack%2Fpuppet-gnocchi~master~If54dff68ad1c03e07e66399b0dc25f0257c05f58","project":"openstack/puppet-gnocchi","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"If54dff68ad1c03e07e66399b0dc25f0257c05f58","subject":"WIP: Convert policy.json to policy.yaml","status":"ABANDONED","created":"2021-03-24 13:44:06.000000000","updated":"2021-03-25 10:11:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7aedf48fcb64b12147b396fa3ba504a7b4bc3971","_number":782743,"virtual_id_number":782743,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-25 00:02:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 14:05:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"299239d9c8fc4f1933e44cd80f890ae9d2cdb6d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 13:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee8ed131b6441302ac4f404b9f0ff852244c9405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:05:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f9b6b0727d74dc985ab0f8ddf1e2e2a : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfe46c35fb2a47dd8cffa87231701382 : FAILURE in 4m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/91a9f99015d747ed9a111b507c92ea01 : FAILURE in 11m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/209734a754aa42828655ad122165da79 : FAILURE in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b2db5403f8949ffb1c63980d37e26d6 : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6975788028c34781832b92aec822ebaf : FAILURE in 9m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f228beb6f27b43d1befaec02ad823154 : FAILURE in 13m 33s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6e6d295a03e4c83ae3cd82d4afa5c47 : FAILURE in 10m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6f89f526d5342f7afae41ab38d9d712 : FAILURE in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"548a73ebcd11224310513cf67d2e18012acbfb08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 14:22:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7323e398e328a35b6b17d723ebd0a2473e66496b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:42:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/951be07189664e97a1fb1b866c2f7de2 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cb2fdcfea4d40b3b7aa9e18ecf7317a : SUCCESS in 4m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2049b9e9410b4c3d9144c7fc8f909cc5 : FAILURE in 9m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bd783d902cb42a88a70914c1bda96e7 : FAILURE in 8m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc59341b84de40f09f228aebaab64d08 : FAILURE in 7m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3816ff18e5d943e59608ed8757b5ef11 : FAILURE in 7m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5dd79c47666648098235175842f4b997 : FAILURE in 13m 02s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8d9c65671e74b71a3f8ab80a5a8a66e : FAILURE in 11m 01s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/595f508925c64455b07801aea9ad2ca4 : FAILURE in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09a003f2c7c132a659443418a036ba03861f45cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 14:59:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7592fb50a74adedfc18c1c204d4c71070f4aefea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 15:54:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/159b84acfe6a4052aeceb952a885325f : FAILURE in 5m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccadf29048f041089404b3ac0894f454 : SUCCESS in 5m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad9e0233e7794eff8da9847b552c8926 : SUCCESS in 11m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64b1b992717c47a9a8ef47c5a82b3e45 : SUCCESS in 8m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8407e016206d457eaf423d4ab8597d0d : SUCCESS in 8m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c027b88d57324de6978497310258c962 : FAILURE in 34m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/546db36a9a5c4e4d93714509f1b44952 : FAILURE in 43m 36s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d3f1badd66142b5a985a23e17189383 : FAILURE in 17m 37s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbdab580ac26422fb6fdcbd819335693 : FAILURE in 24m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c3e77887f80b9fa8f53451c7eab55b1a9ff354e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 22:54:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cea39cc40e65b052268368ee30c54dc935684a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:02:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fd7e86d6e294928897eefd207e969c9 : FAILURE in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12ea836c8cbf41b6bae4510e4a1ad614 : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9432848a682420589598af9129187a8 : SUCCESS in 9m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af821d4a80f04c0c942f96a4ac20eefe : SUCCESS in 7m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/709bbe6e1d7d49faa1a2b47b1deb5183 : SUCCESS in 7m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2abe8cf363d14d32987d151a01b85c5a : SUCCESS in 49m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2831a6aa4e24fe2b2899b25396f9a3f : SUCCESS in 1h 06m 01s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdf1151eb051427192bcce8a961b906a : SUCCESS in 16m 52s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b1e4dcec7c14599811e38dfedc32da6 : FAILURE in 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7aedf48fcb64b12147b396fa3ba504a7b4bc3971","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-25 10:11:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9e6e0ac3d488fbaa424cd1cd741a4c30367c86b","revisions":{"5c4f326da94addf3be6363a02fadbe275b11a70e":{"kind":"REWORK","_number":1,"created":"2021-03-24 13:44:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/782743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/43/782743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/1"}}},"commit":{"parents":[{"commit":"d7c49fc87cf6b216f84628a4302644a7fb5ec803","subject":"Use yaml instead of json for policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d7c49fc87cf6b216f84628a4302644a7fb5ec803"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 13:43:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 13:43:20.000000000","tz":540},"subject":"WIP: Convert policy.json to policy.yaml","message":"WIP: Convert policy.json to policy.yaml\n\nChange-Id: If54dff68ad1c03e07e66399b0dc25f0257c05f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5c4f326da94addf3be6363a02fadbe275b11a70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5c4f326da94addf3be6363a02fadbe275b11a70e"}]},"branch":"refs/heads/master"},"43afbdf09fff3d2b4fdca1d29f87db6cc4c8c68a":{"kind":"REWORK","_number":2,"created":"2021-03-24 14:22:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/782743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/43/782743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/2"}}},"commit":{"parents":[{"commit":"d7c49fc87cf6b216f84628a4302644a7fb5ec803","subject":"Use yaml instead of json for policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d7c49fc87cf6b216f84628a4302644a7fb5ec803"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 13:43:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 14:22:12.000000000","tz":540},"subject":"WIP: Convert policy.json to policy.yaml","message":"WIP: Convert policy.json to policy.yaml\n\nChange-Id: If54dff68ad1c03e07e66399b0dc25f0257c05f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/43afbdf09fff3d2b4fdca1d29f87db6cc4c8c68a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/43afbdf09fff3d2b4fdca1d29f87db6cc4c8c68a"}]},"branch":"refs/heads/master"},"1430136d3d7d076e3a765199ca9d80c18ddd0af0":{"kind":"REWORK","_number":3,"created":"2021-03-24 14:59:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/782743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/43/782743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/3"}}},"commit":{"parents":[{"commit":"d7c49fc87cf6b216f84628a4302644a7fb5ec803","subject":"Use yaml instead of json for policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d7c49fc87cf6b216f84628a4302644a7fb5ec803"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 13:43:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 14:59:17.000000000","tz":540},"subject":"WIP: Convert policy.json to policy.yaml","message":"WIP: Convert policy.json to policy.yaml\n\nChange-Id: If54dff68ad1c03e07e66399b0dc25f0257c05f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/1430136d3d7d076e3a765199ca9d80c18ddd0af0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/1430136d3d7d076e3a765199ca9d80c18ddd0af0"}]},"branch":"refs/heads/master"},"b9e6e0ac3d488fbaa424cd1cd741a4c30367c86b":{"kind":"REWORK","_number":4,"created":"2021-03-24 22:54:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/782743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/43/782743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/43/782743/4"}}},"commit":{"parents":[{"commit":"d7c49fc87cf6b216f84628a4302644a7fb5ec803","subject":"Use yaml instead of json for policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d7c49fc87cf6b216f84628a4302644a7fb5ec803"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 13:43:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 16:30:26.000000000","tz":540},"subject":"WIP: Convert policy.json to policy.yaml","message":"WIP: Convert policy.json to policy.yaml\n\nChange-Id: If54dff68ad1c03e07e66399b0dc25f0257c05f58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b9e6e0ac3d488fbaa424cd1cd741a4c30367c86b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b9e6e0ac3d488fbaa424cd1cd741a4c30367c86b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
