)]}'
{"id":"openstack%2Fpuppet-gnocchi~785237","triplet_id":"openstack%2Fpuppet-gnocchi~master~Iea9a674be86902be46fdc124e6d0ecf0c1ec8961","project":"openstack/puppet-gnocchi","branch":"master","topic":"debian-uwsgi-support","hashtags":[],"change_id":"Iea9a674be86902be46fdc124e6d0ecf0c1ec8961","subject":"Add support for gnocchi_api_uwsgi_config in Debian","status":"MERGED","created":"2021-04-07 16:55:58.000000000","updated":"2021-04-14 16:21:20.000000000","submitted":"2021-04-14 16:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785237-debian-uwsgi-support","meta_rev_id":"ce26a43aba2223ad635e8491210cb812a9b0fa7b","_number":785237,"virtual_id_number":785237,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 16:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-04-14 14:17:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-14 14:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-04-14 14:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 18:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 06:47:46.000000000","updated_by":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"reviewer":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"state":"REVIEWER"},{"updated":"2021-04-12 06:55:37.000000000","updated_by":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"reviewer":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"state":"REVIEWER"},{"updated":"2021-04-12 07:16:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-04-12 07:16:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-04-12 07:16:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-12 07:16:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"63c3eb65741832cb170bb528e76eab862acb20b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 16:55:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10f19bd876c601c32801d0b94ebba8268a63fde6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 17:02:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78c58075c099976661990776da696bafa5a4936d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 18:31:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29a3808e51314f45b1ae0ca176b2f891 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74aa554e2d1747d49239f68631325bf2 : SUCCESS in 4m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4de04b9653034df1b5325afa075ffa90 : SUCCESS in 12m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37553e10799a492b9e0338cc02228204 : SUCCESS in 7m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efbaeddc7efb426d936236c59f64fae2 : SUCCESS in 6m 09s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96d49f5f225f4964a61da6386c5caa3c : FAILURE in 36m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/138160600f7840f69ecd476a7ab839ba : SUCCESS in 1h 20m 00s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b37d06abb4814abd901cc4de2984911f : SUCCESS in 19m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7729b88e2ee44a4dbb83877e150f865e : FAILURE in 23m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d56a3e6318ec48339405604c8e3457f7 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"0e020eead82db3e0aed30820a5dc330767546672","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2021-04-12 06:47:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"37e58c9503753e87c41add55ef5346378c0dc3b9","author":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"date":"2021-04-12 06:55:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"88a89cb8bee2668e133d1932098f7233cb8294ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 16:21:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a81b6fb4067cef385c6ba8da7f52eaeda45e371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 17:15:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"17a0840ec560040b2953ef5bf82fb7f6d25fa2df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 17:40:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a1d4032d70396b6c974724659634c1a0a66e6070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 18:56:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d123705b90e404d88d0b10fa1d2006c : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f75332494f345aca042acf80b5e43a2 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2f32e903ec14c25983fa5c4c67b2ce8 : FAILURE in 11m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/885e6449e01e46bab1c9ea395af193bf : FAILURE in 8m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0a830a8d38341389fb130ad3b043479 : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c78a4e62a8f44fe5afe5232b31edf711 : SUCCESS in 47m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/acec87137abd46ed865e93b42789daef : FAILURE in 1h 13m 59s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e065f1ab6124b829495529de5c2d675 : SUCCESS in 17m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8b3c91b13e04a619621878874f08236 : FAILURE in 24m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6c8462c3fdb4f1bb608a3665cb0a72b : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":5},{"id":"de6d155069bab9858d19ffb50a5019d55cd4822e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 19:06:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bef19fd3883caff504ba1af4214ee523f8c996af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 19:58:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"104fcacabc4bd893e75bafdb4d65ad65695958a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 21:13:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d2941e5fcbf727a216bd85bfaa5fb21561f1e230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 22:30:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4845c971f3642fdb11c4413a31582fe : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d707cdd1db8d472689bf1a01bfca4a65 : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/06b21ec78310457cb6af699d70231623 : SUCCESS in 9m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cda17d608064146abb5961ebf2167ee : SUCCESS in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee1bd9d325ac43848bb3e051e805d9c4 : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e44d0515b924a67a194b8b757e252a6 : SUCCESS in 50m 50s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a9fc7ecf09241fc819343ec5430eb99 : SUCCESS in 1h 12m 22s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c38e9ddbd27e4bbd8d7e0c6a15b62fd6 : SUCCESS in 16m 38s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8c901fcd8f34d7b8ed48767fcc54b6b : FAILURE in 21m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fd309ae0e034c9abbd075cce9032039 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":8},{"id":"9df602a2d225d550a8fbbe13e3c79732d138b2a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-14 14:17:38.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"492ec365d50dc9e82c7dad242456356d16a2eb84","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-14 14:35:17.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"f96f45fb8a1f689a0932008593b5dda639e42904","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:35:31.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c382379c73bb2b9119349b100a4b121b0888fa4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:19:28.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fcf66da57b2463498fb54cf5149eda8 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02bf32e1e32d4df8a9c07ee3d6a57890 : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f382d4379dc04c228ab41763b5bb6b43 : SUCCESS in 9m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f5e4e3879d649ed94c6944650366768 : SUCCESS in 7m 48s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/188ebd0a942e48e18027a7334ca03236 : SUCCESS in 1h 38m 19s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2a3a45c585f4f63897adc0f23052cfb : SUCCESS in 20m 26s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ded79f76e094c78a520e99e4764eaca : FAILURE in 21m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0171b520b2cf48cb8c09282ef1af08ff : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":8},{"id":"b082d9e0ad153a36e432d4db35d38a40063692f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:19:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"ce26a43aba2223ad635e8491210cb812a9b0fa7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:21:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/912293a758bb4b51a4333193322691e1 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8dcef88edb5b9883c23510be8a2ab67d9e45b576","revisions":{"05237c57075f635c9b93661b4ee66a66b485b32e":{"kind":"REWORK","_number":1,"created":"2021-04-07 16:55:58.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/1"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/05237c57075f635c9b93661b4ee66a66b485b32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/05237c57075f635c9b93661b4ee66a66b485b32e"}]},"branch":"refs/heads/master"},"5166893bab2cf7706e44d3eeda48133c8e44263b":{"kind":"REWORK","_number":2,"created":"2021-04-07 17:02:47.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/2"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 17:02:32.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5166893bab2cf7706e44d3eeda48133c8e44263b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5166893bab2cf7706e44d3eeda48133c8e44263b"}]},"branch":"refs/heads/master"},"543c0d3d145e6ebbb70af9de682f630aa0507ad9":{"kind":"REWORK","_number":3,"created":"2021-04-12 16:21:24.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/3"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 16:21:13.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/543c0d3d145e6ebbb70af9de682f630aa0507ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/543c0d3d145e6ebbb70af9de682f630aa0507ad9"}]},"branch":"refs/heads/master"},"a5c135e8b1d484deb2fe69c88eb65ea86c886837":{"kind":"REWORK","_number":4,"created":"2021-04-12 17:15:10.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/4"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 17:14:57.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a5c135e8b1d484deb2fe69c88eb65ea86c886837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a5c135e8b1d484deb2fe69c88eb65ea86c886837"}]},"branch":"refs/heads/master"},"886caabcb3f695376b5bd3a9ae1322b8926b0512":{"kind":"REWORK","_number":5,"created":"2021-04-12 17:40:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/5"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 17:40:06.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/886caabcb3f695376b5bd3a9ae1322b8926b0512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/886caabcb3f695376b5bd3a9ae1322b8926b0512"}]},"branch":"refs/heads/master"},"b792f6b4472fc1434543ba73948ccd49b484aacf":{"kind":"REWORK","_number":6,"created":"2021-04-12 19:06:49.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/6"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 19:06:38.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b792f6b4472fc1434543ba73948ccd49b484aacf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b792f6b4472fc1434543ba73948ccd49b484aacf"}]},"branch":"refs/heads/master"},"d261ac09ed796861c71c1499ba320b59eae1f476":{"kind":"REWORK","_number":7,"created":"2021-04-12 19:58:07.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/7"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 19:57:55.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d261ac09ed796861c71c1499ba320b59eae1f476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d261ac09ed796861c71c1499ba320b59eae1f476"}]},"branch":"refs/heads/master"},"8dcef88edb5b9883c23510be8a2ab67d9e45b576":{"kind":"REWORK","_number":8,"created":"2021-04-12 21:13:46.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/37/785237/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/37/785237/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/37/785237/8"}}},"commit":{"parents":[{"commit":"620df2f563a18563c71c4320dc4782a967e33e31","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/620df2f563a18563c71c4320dc4782a967e33e31"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 16:54:22.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 21:13:34.000000000","tz":120},"subject":"Add support for gnocchi_api_uwsgi_config in Debian","message":"Add support for gnocchi_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Gnocchi API. Therefore, this patch adds a new\ngnocchi_api_uwsgi_config providers as well as a new\ngnocchi::wsgi::uwsgi_api class.\n\nChange-Id: Iea9a674be86902be46fdc124e6d0ecf0c1ec8961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/8dcef88edb5b9883c23510be8a2ab67d9e45b576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/8dcef88edb5b9883c23510be8a2ab67d9e45b576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
