)]}'
{"id":"openstack%2Fpuppet-gnocchi~786807","triplet_id":"openstack%2Fpuppet-gnocchi~master~I2cfa1267b08ea432a8bf3245ebb57abc5fadb809","project":"openstack/puppet-gnocchi","branch":"master","topic":"enforce_new_defaults","hashtags":[],"change_id":"I2cfa1267b08ea432a8bf3245ebb57abc5fadb809","subject":"Add support for oslo_policy/enforce_new_defaults","status":"MERGED","created":"2021-04-19 00:31:40.000000000","updated":"2021-04-27 10:54:07.000000000","submitted":"2021-04-27 10:51:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786807-enforce_new_defaults","meta_rev_id":"00facb5d63b3d60766a04f02f2a3811e0e95eff5","_number":786807,"virtual_id_number":786807,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 10:51:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-04-19 21:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-04-27 09:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-04-27 09:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-19 02:05:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 21:19:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-27 09:01:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"12b0aadbb4ef414f6d6c9e694f464b8734aedf5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-19 00:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90a38693e8a69943e170d8b1f9cbc772f8cba86a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 02:05:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ac15fd94920494b926b9ce85d2de13e : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfa0f23d3599449bacf4f9999024684d : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c12a10a41b28431b9d34e1fc4da2c08b : SUCCESS in 8m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75df68b9b97244ee9eb6c401eff782fe : SUCCESS in 7m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cfc013a37f849ffadcbb8c63279de79 : SUCCESS in 7m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1efe75e806ca4bfe9a63e4722ecff5c7 : FAILURE in 33m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a0ffe2cf2414afcb3c86ee0e02aa1b2 : SUCCESS in 1h 28m 15s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d477c2fb3ba2446a9109cb5a65a7f170 : SUCCESS in 17m 19s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60893b6302284d1685ff1a086a66767d : FAILURE in 22m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/509e8960e18a405eba1ee32e361bf8e1 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"91d7d0e04d34bb9cfebf19adf1e13f7dadbdbe0c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-19 21:19:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d12f964475d695462956c5768a7e74bf3e27417d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-27 09:01:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8adf8d7c197f8acf99cc1f28127a3eab2d9a2899","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 09:03:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b28a4cd1a0d63795c09180825a034453973f00fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 10:51:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d30cd35621f943a2b9b76cdf14589a3d : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d349cdee794e179a6bd5ae35014550 : SUCCESS in 4m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e754411954fa44a3a6b0fa69f3999165 : SUCCESS in 11m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c607bf9562d4db489fd5ffa123c6604 : SUCCESS in 8m 34s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/52133dfe07504ca5b4aa0398e003ce84 : SUCCESS in 1h 11m 32s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f11b31723b043d89fc756cd9eda875a : SUCCESS in 20m 17s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cf48da22b0143ef9b71c250e114e3ce : FAILURE in 21m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d95f790e003c45739ffdb46f76be7625 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e8e9d8219bf932adb32b0203325b75ac3833982c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 10:51:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00facb5d63b3d60766a04f02f2a3811e0e95eff5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 10:54:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d1fae7a5a4e4841ad908b55399f185b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e1d398f56a251d0c77a6d6a133a9a10abcf3414","revisions":{"0e1d398f56a251d0c77a6d6a133a9a10abcf3414":{"kind":"REWORK","_number":1,"created":"2021-04-19 00:31:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/786807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/07/786807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/786807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/786807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/786807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/786807/1"}}},"commit":{"parents":[{"commit":"14e1b1027b3d3b3968611869c4910b5db3399874","subject":"Add support for the oslo_policy/enforce_scope parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/14e1b1027b3d3b3968611869c4910b5db3399874"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-19 00:31:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-19 00:31:22.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nDepends-on: https://review.opendev.org/781428\nChange-Id: I2cfa1267b08ea432a8bf3245ebb57abc5fadb809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0e1d398f56a251d0c77a6d6a133a9a10abcf3414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0e1d398f56a251d0c77a6d6a133a9a10abcf3414"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
