)]}'
{"id":"openstack%2Fpuppet-gnocchi~787556","triplet_id":"openstack%2Fpuppet-gnocchi~master~Ib674673edec143841aa88d754ea2e9b0cb13be1a","project":"openstack/puppet-gnocchi","branch":"master","topic":"init-settings-no-openstack-settings","hashtags":[],"change_id":"Ib674673edec143841aa88d754ea2e9b0cb13be1a","subject":"Debian api_uwsgi_config: Switch to :ini_setting","status":"MERGED","created":"2021-04-22 13:20:09.000000000","updated":"2021-04-26 14:53:45.000000000","submitted":"2021-04-26 14:53:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787556-init-settings-no-openstack-settings","meta_rev_id":"83b3bbfdde49071cd59400c2633f551ec900a64d","_number":787556,"virtual_id_number":787556,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 14:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-04-26 13:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-22 15:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-04-26 13:12:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 14:41:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 15:16:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-26 13:12:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"4a934b82ffc5d2ae70d450b0c990dc3618700c5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-22 13:20:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"095237e9949ebd4fe1bbd97936fbecf13466fa89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 14:41:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f418616fe9c48608e11e212069f347b : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17b79f3330c540aeb83d775014ad38ca : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cdd59356d994a7bb89168c31f305460 : SUCCESS in 12m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43adc01b4009498d98b95fd0deec35fa : SUCCESS in 8m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac7b9e7f49ae4cd8bf551b95ec49625b : SUCCESS in 9m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84ac17ee5bc74e209cb8c4f67fde6ba3 : FAILURE in 41m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/23d57d9116a347f6835ffb7a772c291f : SUCCESS in 1h 17m 23s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/81920a4c61194e56ade769b94f470351 : FAILURE in 22m 02s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d480e296cf6d4bbea71e4458ba908d96 : FAILURE in 24m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc38b6e12fe70d209151de12828966f55a95a2a9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-22 15:16:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"505a1da26d4c94a5533bdd0106569368f964a1c9","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-22 21:34:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dbb00a413bc006d98623f51f965c70762a3f054c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 22:42:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75ad12b21fb94f77a75f21f6371d5f50 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f4548e4e67044ec861f9afba0790682 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3b3b072b4cf499a9e1795379ca1d1d1 : SUCCESS in 15m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19a40acfc4664c2db0a7fb1bcf905716 : SUCCESS in 10m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/857af22b147a48a6b56340190039b873 : SUCCESS in 6m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56fd222ddb8d4ae08588d8d1bc324eda : SUCCESS in 48m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cb4a6472b9c4904bf08dc8924672b3a : SUCCESS in 1h 05m 36s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6aafd04867514c72a20567d62089f5ed : SUCCESS in 21m 00s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b66813140c241b98ac41e74771f5997 : FAILURE in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b685cc2a3075d61824ebadd485dc09b29d2a9568","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 13:12:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baf47426252838677e0d59b2bc0502a3d81be5cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 13:12:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ea4317a1ce96ac1b941a8fcd7073e7a24a2224d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:53:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9120cd1a37e419b977e49110c98ca83 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e1596aac7d141bca8672617df55b765 : SUCCESS in 3m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/59813b164379459da1fc99543f1e7364 : SUCCESS in 9m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68b9ef6fb51e4eed8f6e54136e6fad3a : SUCCESS in 7m 35s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e7f3c30e1174970946fc6a3c4cf2ab5 : SUCCESS in 1h 12m 31s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6db8d78c638c4fd5932721284f9d0bfd : SUCCESS in 17m 30s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e25c78930ad4587960287af787a6995 : FAILURE in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83b3bbfdde49071cd59400c2633f551ec900a64d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:53:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e986276b7751a99aadd9dd02be9d2a9ed270f82d","revisions":{"e986276b7751a99aadd9dd02be9d2a9ed270f82d":{"kind":"REWORK","_number":1,"created":"2021-04-22 13:20:09.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/56/787556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/56/787556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/787556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/787556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/787556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/787556/1"}}},"commit":{"parents":[{"commit":"4245a9e82380c709ff6399781a116de502f51a79","subject":"Merge \"Add support for gnocchi_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/4245a9e82380c709ff6399781a116de502f51a79"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-22 13:19:42.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-22 13:19:42.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Ib674673edec143841aa88d754ea2e9b0cb13be1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/e986276b7751a99aadd9dd02be9d2a9ed270f82d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/e986276b7751a99aadd9dd02be9d2a9ed270f82d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
