)]}'
{"id":"openstack%2Fpuppet-gnocchi~792832","triplet_id":"openstack%2Fpuppet-gnocchi~master~Ibf79b167a9d193974180949c95557d77ca495499","project":"openstack/puppet-gnocchi","branch":"master","topic":"file-incoming","hashtags":[],"change_id":"Ibf79b167a9d193974180949c95557d77ca495499","subject":"Add support for the file incoming driver","status":"MERGED","created":"2021-05-24 16:30:26.000000000","updated":"2021-06-28 10:37:33.000000000","submitted":"2021-06-28 10:36:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792832-file-incoming","meta_rev_id":"788ea9ad8c076fe48b94b9f168f92c339cbe7782","_number":792832,"virtual_id_number":792832,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 10:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-05-25 16:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-06-28 08:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-06-28 08:59:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 18:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-25 16:07:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-28 08:59:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0a748cc9a0482d28740aba342ee9794457c6ec9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-24 16:30:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb92e50ea8c571f78fc126e27013eda7e2c4c751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 18:27:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dacbb681037341d09f0ef191f7795145 : FAILURE in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a73df2699cd4c089b0fae939e27f233 : FAILURE in 4m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ec2d2a3d04d4d659c3cf4c9841882d1 : FAILURE in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b63a6c42a6e49e2b228d9440da75b90 : FAILURE in 9m 48s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5c6dce34e744a478a09378a2b06e4cf : FAILURE in 7m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2920f5b267e403894ac62b5e4717dc8 : FAILURE in 6m 51s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f5070f08edb4784ad75bd98792380f2 : FAILURE in 6m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/889b5cb8a97246ad875f71e8daff3211 : FAILURE in 6m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/964969b1318545b19aa3bf6bd6e9d83e : FAILURE in 51m 22s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/37d6ca655b7e4918a7e3b507a2dc6627 : SUCCESS in 1h 35m 11s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/244d9cb9ee0b427ea755c32339828b22 : SUCCESS in 1h 09m 26s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7a289e5cbab4c63aaae3f830ae39f57 : SUCCESS in 18m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f5e4b2ab8f54ec89eed35bd421f2032 : FAILURE in 17m 48s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d069a349b704d2e885fe884888ef594 : SUCCESS in 20m 39s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51ed31c97fdf4976aefc81bbba54c189 : FAILURE in 22m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fb1064425c74b23ba590ae1cad253c6 : TIMED_OUT in 1h 32m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d237293f58fd4dd9bc4109d9939ca852 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9fcc7663d7b15609637d3ff13150e0335a83d3b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-24 23:02:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54f5a2d16fcd8caac492a75aa2355b7ba618a3ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-24 23:20:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0720a400457ccd3f051d2da5cdf869d8ed3b8182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 23:58:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/545eeae9272045fe9c350cd41f1490c5 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9769599062d24ef880693c753e7fdb2d : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5375c9ff24ba412d9d6db75f4f1cfd6c : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6044bc8980f401bbebd76bf514d6822 : RETRY_LIMIT in 4m 36s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/92d0fb499166455baab92aa64e0406e9 : RETRY_LIMIT in 5m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f4d523f3cc94b899f417b35665c14c7 : SUCCESS in 6m 44s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33e574e10b3e4c3f9a9b8f2408daa607 : SUCCESS in 7m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41034ab1cf7649839654b0d96154f518 : SUCCESS in 7m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e41a47a859e4476851d1c6a42a38f56 : FAILURE in 36m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb964bdb797e4180a2d46147c13cb057 : RETRY_LIMIT in 4m 50s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5ac7dc2f01d468094871404c6a4f3f3 : RETRY_LIMIT in 6m 28s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/78ae6b1531ac409b8522664b51415191 : RETRY_LIMIT in 5m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5f2989d1b33420eaab2fec09aea403d : FAILURE in 14m 13s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e6150aabc3847ff85d68106a32d5297 : RETRY_LIMIT in 5m 12s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35758db83856414ea7fe22aeecb1e6d9 : FAILURE in 22m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/033436e6f75c49bfb7bcc4f0d2bfa485 : FAILURE in 21m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a9aac5f1c3d47fd9ad591d059dd34be : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"801b80d0b4c368ff3943422f53886a7f6be23f85","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-25 13:34:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9430b8411bc70b1d655aff925c9ba8a61f3b632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 14:56:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1101ec4a588f44f0be447686397b951b : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba5af7dc41204c03b7aa82eae6d3315d : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ad974f9e41458d9b63aa39668f4197 : SUCCESS in 4m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/def71aa376524532bdaa14fd20a931f6 : SUCCESS in 11m 01s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/563344fa131e47ba8df3c854948f8f92 : SUCCESS in 10m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e7b78daf60346359ff641cabaf03083 : SUCCESS in 10m 14s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3af5f140aba34dc1bab8cddad6ca103b : SUCCESS in 9m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edcbecfa65024abb84799747ea504829 : SUCCESS in 8m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c63ac09833a14e02925e794a65226b64 : FAILURE in 41m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6862c1b8606f40da8eab961ec7baa087 : SUCCESS in 1h 11m 51s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/90684f6a7c934dfbb4c2258888e56d01 : SUCCESS in 1h 20m 32s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a7f2b67ebbf4b20a26a1084a4002333 : SUCCESS in 19m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54929232cb264a21afd87b74c8fac6e9 : FAILURE in 15m 16s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2322349223043ae8965e4594c14fbff : SUCCESS in 19m 37s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a1908ec07f74250aa9911d39d14abef : FAILURE in 24m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8cb5b2bc2324bbcbc48d9da46789b34 : FAILURE in 24m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b7a9f8cbdec4ccfb41e992ff72c212c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":3},{"id":"4a9cd7f9df6134e0c296809b5e5acc811e7ca03e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-25 16:07:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e7f235038e90f9b898d08c75d04716f19a470638","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-28 08:59:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"52f2cfe15cbc6190f6f1613123b5b4c58f10a5bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 09:00:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b257a5ce2daa23045aadfb627d151c6aba5be883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 10:36:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d17fcd4a2f5b4618a19092f3c7c2bd8b : SUCCESS in 5m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d25556b6d9c34bdc9e228166dd568ba1 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04947a97d9084b6384b547545054677e : SUCCESS in 6m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2972b2e5849d451c9510c94442346de9 : SUCCESS in 12m 05s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ddc0259853a440c94ec7f28c046b5ec : SUCCESS in 14m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3641936b724641c495bffea08186e51b : SUCCESS in 14m 27s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c35f73b7f6f4bdbab09c638a8b23301 : SUCCESS in 11m 16s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee757c5d61b04e1cb142c16baaa38271 : SUCCESS in 1h 30m 43s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/82c618ecee5644a1a6e4a9907c844fea : SUCCESS in 24m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce2c654e8b324e83ad9430b137bfad5b : FAILURE in 15m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb40ca16776946a389957b006eb7db24 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":3},{"id":"535d0265d8585a243de7381dd3f71af5c886567e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 10:36:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"788ea9ad8c076fe48b94b9f168f92c339cbe7782","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 10:37:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3098f01fa6eb44e0a6e56836944e945a : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a00bfa36426708c5b674d2467e345e0dbcf4f453","revisions":{"66181b86e3a52d0a916e994743d1e472d4523403":{"kind":"REWORK","_number":1,"created":"2021-05-24 16:30:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/792832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/32/792832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/1"}}},"commit":{"parents":[{"commit":"cfec496edd4f5db1de7d92c8579ca8e14f005b8b","subject":"Fix a test case with typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/cfec496edd4f5db1de7d92c8579ca8e14f005b8b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 16:26:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 16:30:17.000000000","tz":540},"subject":"Add support for the file incoming driver","message":"Add support for the file incoming driver\n\nThis change introduces a new gnocchi::storage::incoming::driver class\nso that users can configure the file incoming driver of Gnocchi, which\nstores incoming data in local (or shared) filesystem.\n\nChange-Id: Ibf79b167a9d193974180949c95557d77ca495499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/66181b86e3a52d0a916e994743d1e472d4523403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/66181b86e3a52d0a916e994743d1e472d4523403"}]},"branch":"refs/heads/master"},"bba8c9d645100cbaecc03f61717927cbd09ca2ea":{"kind":"REWORK","_number":2,"created":"2021-05-24 23:02:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/792832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/32/792832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/2"}}},"commit":{"parents":[{"commit":"cfec496edd4f5db1de7d92c8579ca8e14f005b8b","subject":"Fix a test case with typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/cfec496edd4f5db1de7d92c8579ca8e14f005b8b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 16:26:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 23:01:56.000000000","tz":540},"subject":"Add support for the file incoming driver","message":"Add support for the file incoming driver\n\nThis change introduces a new gnocchi::storage::incoming::driver class\nso that users can configure the file incoming driver of Gnocchi, which\nstores incoming data in local (or shared) filesystem.\n\nChange-Id: Ibf79b167a9d193974180949c95557d77ca495499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/bba8c9d645100cbaecc03f61717927cbd09ca2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/bba8c9d645100cbaecc03f61717927cbd09ca2ea"}]},"branch":"refs/heads/master"},"a00bfa36426708c5b674d2467e345e0dbcf4f453":{"kind":"REWORK","_number":3,"created":"2021-05-24 23:20:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/792832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/32/792832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/32/792832/3"}}},"commit":{"parents":[{"commit":"cfec496edd4f5db1de7d92c8579ca8e14f005b8b","subject":"Fix a test case with typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/cfec496edd4f5db1de7d92c8579ca8e14f005b8b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 16:26:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-24 23:19:55.000000000","tz":540},"subject":"Add support for the file incoming driver","message":"Add support for the file incoming driver\n\nThis change introduces a new gnocchi::storage::incoming::driver class\nso that users can configure the file incoming driver of Gnocchi, which\nstores incoming data in local (or shared) filesystem.\n\nChange-Id: Ibf79b167a9d193974180949c95557d77ca495499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a00bfa36426708c5b674d2467e345e0dbcf4f453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a00bfa36426708c5b674d2467e345e0dbcf4f453"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
