)]}'
{"id":"openstack%2Fpuppet-gnocchi~793127","triplet_id":"openstack%2Fpuppet-gnocchi~master~Id1957e4d6e5ad003592ad46e469ec9f9e78ad58d","project":"openstack/puppet-gnocchi","branch":"master","hashtags":[],"change_id":"Id1957e4d6e5ad003592ad46e469ec9f9e78ad58d","subject":"Deprecated manage_cradox and set manage_rados to true","status":"MERGED","created":"2021-05-26 09:25:53.000000000","updated":"2021-06-04 10:46:12.000000000","submitted":"2021-06-04 10:44:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"793127","meta_rev_id":"3351067f70c835588397beddd06cf8e6844377ee","_number":793127,"virtual_id_number":793127,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 10:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-04 10:46:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-04 10:46:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-04 09:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-26 10:56:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 09:23:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-05-27 09:23:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f2a61558f5ce448e7514c3b896800ab335534225","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-26 09:25:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7bab5df9dfc3ab0bbe9819d106bcfb50a62f93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 10:56:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cce42cadaf0741148c2a2cd4b44103d9 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/905e916bbc8845b09af148da3b2889e0 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e9bc1a1188a424fa57909a74a5b150a : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cd714581f2940d4ad01d3ae862bd49e : SUCCESS in 10m 38s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/97b7db09eefe45488892ee1cc977f9f1 : SUCCESS in 10m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e822e69328e4410bb04348ff62f17afc : SUCCESS in 8m 43s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2e8f0174dd64ea4866f00da7e0741ea : SUCCESS in 7m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/250643771c5d41149a9ccc849e57fefe : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cb4568569d04a61aac2f6821bd2516a : FAILURE in 32m 59s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d1b0227e15e41c0a464708cb951876d : SUCCESS in 1h 21m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e8c8e3c7f594fd79add90376fe9cfd6 : SUCCESS in 1h 09m 52s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/99733bdb895047af8df3e77c92d2a9ac : SUCCESS in 17m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb15914e306e4ef48304804351f98673 : FAILURE in 13m 57s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/059f00e4ea2340b3936fccc0f53e31b6 : SUCCESS in 17m 42s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fb6387cf5d04eaa86202b7a15b3951f : FAILURE in 22m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ab614880e9c447e98325fbd17b6c729 : FAILURE in 22m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e96e0cf81111437c85907bb4ce88e95b : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4f1d83856fed89e78e3e5f76cb74b7bd681b6695","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-27 13:16:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"52b884926146e082568963d7b17ba5411bda89e3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-27 13:48:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f96d5a04042b3299ad7dde93881c34e30fac9c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 14:17:07.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c0019e5a857f01239d2a04e8b88216e1b454a099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-05-28 07:50:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55ce7084979a12f79b57534c9b7a83447e146b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:04:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43d5a0deda9f4499919200e3ee19e0dd : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5f3ce0c9807463291b85d3de69c5f02 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/870842ac4ed14640a79d9c696001330e : SUCCESS in 4m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c0632207adc4b689c990c144be5e163 : SUCCESS in 11m 16s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f699d264d3f348eaad8ebe323dfe803d : SUCCESS in 10m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75dcfa842fb3484293fe9f8f007e0226 : SUCCESS in 8m 45s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6370248849f942c7b476bb05fccc38c2 : SUCCESS in 10m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/078272eec72549e680bfabd324a7b2c5 : SUCCESS in 8m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32c0717031a04d2392644dca26cbad6f : FAILURE in 37m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b9795af071344a59c0d47f47e823775 : SUCCESS in 1h 04m 16s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb31eff43c8f4041bf3590f0fda8c228 : SUCCESS in 1h 11m 14s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/31035a95b12c471db01e7245d91201ce : SUCCESS in 18m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9e8fd8518b74427a615f0529179a14e : FAILURE in 16m 38s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3307e6b369e143349102d71d4186ea83 : SUCCESS in 18m 22s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a54be2cfb28d4bccb96b7e2aa5584157 : FAILURE in 24m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/893490d6b549454e91fcca7ff1d8beaa : FAILURE in 24m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbd32722e3864d7aa73032d6b89fcbb3 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9fa33d3a7efd6b8b6042caabc1b0df93eed215c8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-28 11:59:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"26c70f486c90aa9534c3c8e28287fdf112f4fe66","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-01 13:38:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8bcbeaa4cf83baf4b22c95e6f79c4f7d9b985ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 13:40:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e760f09e2c56db547133dc4561ad342defa94faa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 14:46:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ecd37762f63490a864042e5f5dfa7b0 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b294b30a0db46aea80e7cbbfb500549 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cccc779ff444be582b39334db2c74c3 : SUCCESS in 4m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c898a93323ce46abb79eeb97ea66f273 : FAILURE in 10m 36s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/68590b5dc0ca4de1ba27c2d040b1d9cc : FAILURE in 9m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e5cba978db040479712bab9b958889a : SUCCESS in 8m 38s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1c1144463d84a1e85860492773d7efb : SUCCESS in 12m 19s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8061e07b6ddf4643a296ff86d9e8d901 : SUCCESS in 59m 31s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/01d85c3e2d144749846ba1f7186a05ab : SUCCESS in 14m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/084c3791090d41a6b4417812a289efe2 : FAILURE in 20m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a7933ad86e249be809893c1b5d70355 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"eb417a553e250c858e661f6f9a4144fdf17b04a9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-04 07:35:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4d5d1f9a7def772f3244058f54d99e94d3419ecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-04 07:35:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"26cd2f57b7d687b7400d10faacb2505d417548d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 09:12:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5769c2d7310e44d692c29e4e545a7767 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47fae5b44aa848d3933c469a6ccb650b : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9391d2d6e24945c283588d3c0fd86377 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e03bd78e020e48b3afef3d8242c84e76 : SUCCESS in 11m 14s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee41d94e19124111a2736ab53ec49d87 : SUCCESS in 11m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d46c79a65c1f482fa01b2d489726be70 : SUCCESS in 8m 02s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4273c964ccd94b1e80ac4a0ed862427e : SUCCESS in 9m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/802640092336449f89e6701a3b0530dd : SUCCESS in 7m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7d5db4b05c94edcab436deea3ceb86f : FAILURE in 35m 47s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f3045c490f44067ae656504a6badc21 : SUCCESS in 1h 27m 14s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/14e70825c4e0420eae26ec34f77d5aed : SUCCESS in 1h 14m 19s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d92a6a1b86de4934917f2b09587f40ec : SUCCESS in 21m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b91879fb2fba432f850fcd48b834ddba : FAILURE in 13m 38s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a558851acb242e593f0495ddc4330fa : SUCCESS in 18m 44s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21d35184984044a3adb8302358f8a0a4 : FAILURE in 22m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/409be0fa263242008610efc89d81d0a1 : FAILURE in 22m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b09e72885a0548779dc1fd58c7ef3528 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"03a385be00a8049347cb66a4e27bdc8bebbb120d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-04 09:13:36.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0ac1d2e577ecfabc1ebd83bdd8bd41520c3423f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 09:14:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d2a64e05ce03912bc3d299369087e69cfb7b38f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 10:44:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb5208464800449f900e84904d9afb30 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a63659d04d541a88917a44676afa19f : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/649cede204974927895bcbf39b9cbc90 : SUCCESS in 3m 49s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e5272ebb3424ae592e4520cd59b1d2d : SUCCESS in 11m 14s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/485f7d7c7b2d4516baf903101cd7312f : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7516c556c04e47b9a355f5d9bd1411ad : SUCCESS in 7m 53s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/373395481e1d4baa95010b6eb12aec2b : SUCCESS in 7m 57s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc55f2dc2a4b435cae2884065044c9c5 : SUCCESS in 1h 23m 32s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9568b09c27e4046b9a5aed6795dd56a : SUCCESS in 23m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc54777ae57d4077af9482c8628b4644 : FAILURE in 13m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35111e92e3444eee96facda9cad1af73 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"3265c30a4d1873faa2c5d61cfcb4bb7bee50b473","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 10:44:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fc35a385df034dbabc8d3014192c3ca0e43b8929","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 10:46:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/649e48acf6254e64a4792ba6ec693a8e : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3591bd8acc1ac49feb2f0fa33a45a09e94e08407","revisions":{"7bf768c4c5d395163c06f233fab33dcfb9071962":{"kind":"REWORK","_number":1,"created":"2021-05-26 09:25:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/27/793127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/27/793127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/1"}}},"commit":{"parents":[{"commit":"b97255ab253ae0afd142c478c1a0b645861b3a37","subject":"Simplfy the logic to handle coordination_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b97255ab253ae0afd142c478c1a0b645861b3a37"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-05-26 09:24:09.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-05-26 09:24:09.000000000","tz":120},"subject":"Deprecated manage_cradox and set manage_rados to true","message":"Deprecated manage_cradox and set manage_rados to true\n\nThe [1] commit changed the ordering in Gnocchi so that\nthe rados library is preferred over cradox that is no\nlonger maintained and hasn\u0027t been updated in a long time.\n\nCradox is no longer tested in upstream Gnocchi and support\nwas removed in [2] which is not yet released.\n\nThis patch also changes the default value for manage_rados\nto true so that it\u0027s manage by default since it\u0027s the one\npreferred.\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/d5a782e58fc5b3593672fb9c8ac86fddc0c0dedf#diff-60cbc467d17fe0479d5cb2292f716fd73b200973ae43889f72c51227d311a785\n[2] https://github.com/gnocchixyz/gnocchi/commit/bf37e90edaa693880a6f025b1c9329579a7b9882#diff-60cbc467d17fe0479d5cb2292f716fd73b200973ae43889f72c51227d311a785\n\nChange-Id: Id1957e4d6e5ad003592ad46e469ec9f9e78ad58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7bf768c4c5d395163c06f233fab33dcfb9071962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7bf768c4c5d395163c06f233fab33dcfb9071962"}]},"branch":"refs/heads/master"},"a392bb32d9aabc4e8acffeadb79d3ea4a136938e":{"kind":"REWORK","_number":2,"created":"2021-05-28 07:50:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/27/793127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/27/793127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/2"}}},"commit":{"parents":[{"commit":"4b1c63b143c0b860d3aeb1f24d9f7cda848699b9","subject":"Prepare Xena M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/4b1c63b143c0b860d3aeb1f24d9f7cda848699b9"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-05-26 09:24:09.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-05-28 07:50:47.000000000","tz":120},"subject":"Deprecated manage_cradox and set manage_rados to true","message":"Deprecated manage_cradox and set manage_rados to true\n\nThe [1] commit changed the ordering in Gnocchi so that\nthe rados library is preferred over cradox that is no\nlonger maintained and hasn\u0027t been updated in a long time.\n\nCradox is no longer tested in upstream Gnocchi and support\nwas removed in [2] which is not yet released.\n\nThis patch also changes the default value for manage_rados\nto true so that it\u0027s manage by default since it\u0027s the one\npreferred.\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/d5a782e58fc5b3593672fb9c8ac86fddc0c0dedf#diff-60cbc467d17fe0479d5cb2292f716fd73b200973ae43889f72c51227d311a785\n[2] https://github.com/gnocchixyz/gnocchi/commit/bf37e90edaa693880a6f025b1c9329579a7b9882#diff-60cbc467d17fe0479d5cb2292f716fd73b200973ae43889f72c51227d311a785\n\nChange-Id: Id1957e4d6e5ad003592ad46e469ec9f9e78ad58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a392bb32d9aabc4e8acffeadb79d3ea4a136938e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a392bb32d9aabc4e8acffeadb79d3ea4a136938e"}]},"branch":"refs/heads/master"},"3591bd8acc1ac49feb2f0fa33a45a09e94e08407":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-04 07:35:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/27/793127/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/27/793127/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/793127/3"}}},"commit":{"parents":[{"commit":"064f1554ad6551790e57ce919624d3d6c6d59106","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/064f1554ad6551790e57ce919624d3d6c6d59106"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-05-26 09:24:09.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-06-04 07:35:27.000000000","tz":0},"subject":"Deprecated manage_cradox and set manage_rados to true","message":"Deprecated manage_cradox and set manage_rados to true\n\nThe [1] commit changed the ordering in Gnocchi so that\nthe rados library is preferred over cradox that is no\nlonger maintained and hasn\u0027t been updated in a long time.\n\nCradox is no longer tested in upstream Gnocchi and support\nwas removed in [2] which is not yet released.\n\nThis patch also changes the default value for manage_rados\nto true so that it\u0027s manage by default since it\u0027s the one\npreferred.\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/d5a782e58fc5b3593672fb9c8ac86fddc0c0dedf#diff-60cbc467d17fe0479d5cb2292f716fd73b200973ae43889f72c51227d311a785\n[2] https://github.com/gnocchixyz/gnocchi/commit/bf37e90edaa693880a6f025b1c9329579a7b9882#diff-60cbc467d17fe0479d5cb2292f716fd73b200973ae43889f72c51227d311a785\n\nChange-Id: Id1957e4d6e5ad003592ad46e469ec9f9e78ad58d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/3591bd8acc1ac49feb2f0fa33a45a09e94e08407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/3591bd8acc1ac49feb2f0fa33a45a09e94e08407"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
