)]}'
{"id":"openstack%2Fpuppet-gnocchi~798857","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Fvictoria~I5aa56bfced1301ba6ecc8503d7f8874f1fc67709","project":"openstack/puppet-gnocchi","branch":"stable/victoria","topic":"at-params-stable/victoria","hashtags":[],"change_id":"I5aa56bfced1301ba6ecc8503d7f8874f1fc67709","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2021-06-30 11:29:52.000000000","updated":"2021-06-30 16:04:08.000000000","submitted":"2021-06-30 16:02:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798857-at-params-stable/victoria","cherry_pick_of_change":761071,"cherry_pick_of_patch_set":1,"meta_rev_id":"bffd78991a0c93d383fa4cc1b5314a288cd46111","_number":798857,"virtual_id_number":798857,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 16:02:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-30 14:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-30 14:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-30 14:45:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 11:29:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-30 13:13:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-30 14:45:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"30491e9c22c2918272cab41cbc4edf649158102a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 11:29:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1507ccdf5d612cb6b674c36773a4d0d8792703b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:13:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7996eb551ad54edd8d53de9c0edbd303 : SUCCESS in 6m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe753b6dd55d40b79c8a0f928c388def : SUCCESS in 5m 48s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/84dbcae4d64940f8960e4b2970f9f9f3 : SUCCESS in 14m 09s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa5b8519850b4b7a8120c5133ca63548 : SUCCESS in 11m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/493289c89ab142ee8401707aed62e8cf : SUCCESS in 12m 07s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98330bcf9c6143349e22a6b1ef1329b3 : SUCCESS in 1h 00m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a7b533cb7a2c47e58c315512249ba1b8 : SUCCESS in 1h 32m 38s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/189e7dca6e34494db49af8188b47b943 : SUCCESS in 1h 26m 20s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/03001c1aed3747eb9e0b54c7916293bc : SUCCESS in 22m 17s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/fde80bbd16304beaab8a43644c37ffa8 : SUCCESS in 25m 41s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/911da4524ac94097a96f9a834f2e123a : FAILURE in 24m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef1d7e241cec4a75980ded23cc8e0ac1 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c419a3f9870ee4b31d7b2abd925f8b3751efe7d7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-30 14:18:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b7ddfa4848637c44656f257fdf9633e4cab658cd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-30 14:45:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7cf99e962d4d79ce0af58d88b751a0483fd00d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:46:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e068c91fe7e82c8de927e4c71034e1469798a62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:02:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a5ae1b5ae5e4d5fb5a22fdb929b3b89 : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b045877534b4c2ebeca02bae89cd1ec : SUCCESS in 5m 43s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/821a8b7e20ec4c09b6b83bb379c26688 : SUCCESS in 15m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dee4c412030a4d77a18b4a8eff9eb669 : SUCCESS in 9m 38s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9b6b3f332d648e7a1a4b72305a3eaad : SUCCESS in 50m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9cd43072b88d4ee48d2b9f5d3e211cfb : SUCCESS in 1h 13m 40s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/99d4f354fc224231b352f2aec78a9924 : SUCCESS in 24m 11s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/863dedeaa70d45e79e2529d1f87add82 : SUCCESS in 25m 48s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42984e425105436a8554ea190fc20b9d : FAILURE in 24m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af297effdff240dd984f7d87455869cb : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bc36ad2967823ccaa5d6a213b9f0dca84dbbfc96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:02:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bffd78991a0c93d383fa4cc1b5314a288cd46111","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:04:08.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d45d298c35a46ef83bd9b294752095a : FAILURE in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57974f53296d1f35258c500ce9dac9a74fa36c2e","revisions":{"57974f53296d1f35258c500ce9dac9a74fa36c2e":{"kind":"REWORK","_number":1,"created":"2021-06-30 11:29:52.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/57/798857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/57/798857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/57/798857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/57/798857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/57/798857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/57/798857/1"}}},"commit":{"parents":[{"commit":"b658b7ac2aaf625af4f8b55cdad8683d50bb60ac","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b658b7ac2aaf625af4f8b55cdad8683d50bb60ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:05:58.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-06-30 11:29:52.000000000","tz":0},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I5aa56bfced1301ba6ecc8503d7f8874f1fc67709\n(cherry picked from commit 0e4e1160a635c2b1927bc74b6bd6b617ae4bcd3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/57974f53296d1f35258c500ce9dac9a74fa36c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/57974f53296d1f35258c500ce9dac9a74fa36c2e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
