)]}'
{"id":"openstack%2Fpuppet-gnocchi~819755","triplet_id":"openstack%2Fpuppet-gnocchi~master~Ide71a2bd652db2c9e7c04d0679f507afb3de0a16","project":"openstack/puppet-gnocchi","branch":"master","topic":"paste_config","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:11:43.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ide71a2bd652db2c9e7c04d0679f507afb3de0a16","subject":"Add support for [api] paste_config","status":"MERGED","created":"2021-11-30 07:14:41.000000000","updated":"2021-12-08 02:31:19.000000000","submitted":"2021-12-08 02:22:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819755-paste_config","meta_rev_id":"4201c7b8b5bfc79b9e5b3e52c32a1c0b49d5caec","_number":819755,"virtual_id_number":819755,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 02:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-30 09:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 08:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-30 09:02:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 15:50:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ad0e04ae2ce2f7566886de0807419fa9f9a12eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 07:14:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f1786a1916e1b19a2b9f1934ffad6d86e74be7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 08:42:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f296c5b6aa2b43f2ae4a337f91e649ba : SUCCESS in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b329047c28f477bb46ac1574e1f81ea : SUCCESS in 4m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad85b9c083440f2938e66b0e21208a1 : SUCCESS in 3m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1fe530bed54422e88db75783c752c3a : SUCCESS in 9m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ae83da97fab486989839edb4392f25f : SUCCESS in 9m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d71ba6313d284dfe9d6bc15ccc880bcf : SUCCESS in 9m 54s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea5bf6a0f9014dc7a5281707e70a1c99 : SUCCESS in 7m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51910b6e45ac497f9907d8e4a55a1694 : SUCCESS in 7m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9317c181fa946a18e7a57e98eaae248 : FAILURE in 50m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa14d938c30c47f79e37e5b93363d2f2 : SUCCESS in 1h 22m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32127cde4891454baa9d6f4af51ef517 : SUCCESS in 51m 39s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc30f89d193435a88b1eff6b30ed542 : FAILURE in 13m 29s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/196046427f7042558a510678c1402e4c : SUCCESS in 20m 01s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/634edb949b8f4deb954d8b7c0eb3e835 : FAILURE in 22m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/959d8231d85d4ccbb2828caf59fb7e3b : FAILURE in 23m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc70b604682c47728bb28181b05c6a86 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1cbb93079d3da0a7569c0dc95598fb211c2a3659","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-30 09:02:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77873a45c3cc4e4e3dc81717cd72068155400966","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:50:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c50a1e8fed6f465436d4761c9bbb7a7dec7bc20","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:11:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d4fd5413177fd89eb9af3b8a2992d1c2cb4cd23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:08:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe6b29f125fc45f8a8b854d5b4a1cd1e : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0f457c84fe54573af6cb22fd17d06ef : SUCCESS in 4m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cd3565cfcef4a2cbe260f1297a4ea29 : SUCCESS in 4m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a0ac17d1c9c4ab886765ce8a1a77222 : SUCCESS in 14m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fb28d7771094e5da87635e3fe3f207c : SUCCESS in 16m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7572df5ed7ca4e35ba4d862fa12fa2a0 : SUCCESS in 13m 15s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea69673af8094522b9ab69a53f425cda : SUCCESS in 11m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25e35f387ea64ad3a5e4ff7bc6f337db : SUCCESS in 10m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a8ff146175a4b1481636e149b23335b : FAILURE in 37m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5679c5eb575e4c928a2f43fe3c6c23f7 : SUCCESS in 1h 15m 32s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b539c8acdcb548f7af0a28f50ce26ede : SUCCESS in 53m 41s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c47b59a8aefe4c46ab76cf8dc925d52d : FAILURE in 17m 12s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d08737cb3b65498a9dffec1c9250bb79 : SUCCESS in 19m 12s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeb59e86e58149e2bd09fbf03a9027fc : FAILURE in 22m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32cb87efdb594edba11e39c68fe2a7cc : FAILURE in 25m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f86f069880794860bad6595f94416a6f : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d14263b719e7b90d8f0b1d13fe68e56b69696576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:10:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8973649b2ddc99556143ea681431fa6d44ea97e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:22:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb11bab640d0447fa1d997053bacceea : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c11fbe2fa83a47e18ed47416dea4092f : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a3b9499ff042fba874bd1db1a46392 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7be6b37556064ee3b573d7c383f7057e : SUCCESS in 12m 06s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/57b875fd5c7e4ec1addc37804315ab09 : SUCCESS in 11m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bb2d579078a4eaaa7ed149ccf3bf472 : SUCCESS in 8m 45s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92ba24dffbc2488bbd816a16af2617c1 : SUCCESS in 8m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/16da73ea6fc54cd797f1556c7d1fa4f7 : SUCCESS in 1h 09m 12s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f09d268eef7485ea6d05faf06cedf62 : SUCCESS in 24m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ff962d5ddb8412f8a7ff1ac906b40a3 : FAILURE in 16m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ca46c5b42654aa5bdff9c8f970fadd5 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ec36c24a306571a37098d3d955f36dafa47a1f47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:22:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4201c7b8b5bfc79b9e5b3e52c32a1c0b49d5caec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:31:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2576e807526d428a931cc8e4d993566c : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20efa3dc7575d3ea27a863a6523c9f3c38c54491","revisions":{"20efa3dc7575d3ea27a863a6523c9f3c38c54491":{"kind":"REWORK","_number":1,"created":"2021-11-30 07:14:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/819755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/55/819755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/55/819755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/55/819755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/55/819755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/55/819755/1"}}},"commit":{"parents":[{"commit":"34d63f505a5fe18c91330045289306d294dd9b5e","subject":"Allow arbitrary settings of api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/34d63f505a5fe18c91330045289306d294dd9b5e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 07:14:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 07:14:23.000000000","tz":540},"subject":"Add support for [api] paste_config","message":"Add support for [api] paste_config\n\nChange-Id: Ide71a2bd652db2c9e7c04d0679f507afb3de0a16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/20efa3dc7575d3ea27a863a6523c9f3c38c54491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/20efa3dc7575d3ea27a863a6523c9f3c38c54491"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
