)]}'
{"id":"openstack%2Fpuppet-gnocchi~819830","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Fxena~I63bf29acd5ddd07021f8dcaad17bdd3a12783095","project":"openstack/puppet-gnocchi","branch":"stable/xena","topic":"bug/1821580-stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:11:33.000000000","reason":"Takashi Kajinami replied on the change"},"7130":{"account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"last_update":"2021-12-08 01:59:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63bf29acd5ddd07021f8dcaad17bdd3a12783095","subject":"Ensure required packages are installed for s3 support.","status":"MERGED","created":"2021-11-30 12:33:43.000000000","updated":"2021-12-08 02:08:45.000000000","submitted":"2021-12-08 01:59:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819830-bug/1821580-stable/xena","cherry_pick_of_change":647639,"cherry_pick_of_patch_set":12,"meta_rev_id":"84dafed5d68dffb754b55573450e8a7c8c82d591","_number":819830,"virtual_id_number":819830,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 01:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-30 13:09:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:50:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"}],"reviewers":{"REVIEWER":[{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 12:33:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"state":"REVIEWER"},{"updated":"2021-11-30 13:09:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-30 14:02:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 15:50:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"72c0b19538f9b8ec5bd375953f1d7b3daf254364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 12:33:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6a572f32b70f9a20ddc9ffb872c972d9d9f47d50","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-30 13:09:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af8ff87182a8c1aa2c9019a1650cf6e3ac2d54bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 14:02:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b08a291db674ecd8f8e8d89f2b03147 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e9b41bf4c2d420dae6daa292375e910 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20f056370924403eb2902fdcca259c30 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/52e4fcaf3dff45a9bac35cd15ad0a1a2 : SUCCESS in 11m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e83710956254f1ca791f51d4ec24365 : SUCCESS in 10m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f0fe39cdc1e423e8f4bfff719367874 : SUCCESS in 8m 22s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cee0ff4683654538a2d0b43035eb89cf : SUCCESS in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cab41667399c42b38aefa562a3ebd920 : SUCCESS in 8m 19s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92c790d8c99046fab540ec437a8f5841 : FAILURE in 1h 17m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8517b1df071540ce96129ff115513b4f : SUCCESS in 1h 21m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/532fac6275884b9eb6a381ee7bee7c5c : FAILURE in 13m 46s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/718abd66bb304876a0863123200a9d29 : SUCCESS in 19m 59s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b1bc9bf048c4c839f8eb5818ff9326b : FAILURE in 22m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f8ee9a8b0114cf3b572cf8d391875c4 : FAILURE in 22m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92632bbcef79498699630099951aed38 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b8512d4675ee4d85c42d46636ceff8d1fe1897cd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:50:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1361a6120f6b5eeb71b94f42a83a555ea9750016","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:11:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16afadcdf40f878c3582912bdf4609b39c3075b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 00:47:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/332bc72428ff42c1967f8e6a49309fc1 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97155760c0994d528a008f53168d41fc : SUCCESS in 4m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d6d508faab44785a297f5fa0a093b1c : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/74bba439cada42958d6485f54aa3a60d : SUCCESS in 11m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/47a78e4a22834055b930a84f6509970a : SUCCESS in 13m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a6b1413ae9b4552b21a052b0338e735 : SUCCESS in 9m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d676937228e4d619a9166f636cf121f : SUCCESS in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a63c3c944d1c486b98d53b72144675e1 : SUCCESS in 9m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/016590bf53174fd096593a81f9971729 : FAILURE in 36m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/df73e41daaab428cb7e29b1ad521ea75 : SUCCESS in 1h 08m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b9fccfd147341d2a0b5e28eb76e93f6 : FAILURE in 19m 38s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/332f80e1c43948459c39b2044056bbe2 : SUCCESS in 22m 20s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2344ec60f1246c981bdd34afe9c6af2 : FAILURE in 24m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/becd089a80e34dec97cedf75b5424299 : FAILURE in 27m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e9a79b04b249eebf9d497caf10309e : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"25fcefe7695ef3c0e5bf383342e3557dbe5140e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 00:48:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a36e7729a58720454f21f267ea292c650fee6438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:59:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5af1c0ae233b42dbb7dc292533a0f6de : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc55428dc574432b98b2b23474ca6933 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6761de4d7514488b9b438b19bbf1749a : SUCCESS in 3m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6be84d906c4b455bbc7c40f35768c29a : SUCCESS in 11m 40s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/00ab5477782d49faa3532a3d2b7550e7 : SUCCESS in 10m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51596e83fd3d4903930eeed21c100c44 : SUCCESS in 8m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/006a8ba7f3cd42b4a6eec6b514d1296c : SUCCESS in 8m 16s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0692b662b5942709e775fc788558c32 : SUCCESS in 1h 09m 33s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa6ecaf7dbb54f22be000ecd2f05d38d : SUCCESS in 18m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a2a32ae94434a6aa4b5bc8562f5492c : FAILURE in 14m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/247c48e1f0d44342a77d30b35f5087e4 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3d33ebe875585c96bafbb8f08a894e389731a864","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 01:59:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84dafed5d68dffb754b55573450e8a7c8c82d591","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:08:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2753c6acf14afe8dc8699db9cd4a27 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a51f532c5074ef4b57c00927e5f9fe46afd8c342","revisions":{"a51f532c5074ef4b57c00927e5f9fe46afd8c342":{"kind":"REWORK","_number":1,"created":"2021-11-30 12:33:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/819830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/30/819830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/30/819830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/30/819830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/30/819830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/30/819830/1"}}},"commit":{"parents":[{"commit":"1cf22a16db85d00fab2abaeb66384552613c9b4a","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/1cf22a16db85d00fab2abaeb66384552613c9b4a"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2019-03-26 00:29:30.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 12:33:43.000000000","tz":0},"subject":"Ensure required packages are installed for s3 support.","message":"Ensure required packages are installed for s3 support.\n\nChange-Id: I63bf29acd5ddd07021f8dcaad17bdd3a12783095\n(cherry picked from commit e3a16caf261054571a873d6e81a90ff589ddd9fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a51f532c5074ef4b57c00927e5f9fe46afd8c342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a51f532c5074ef4b57c00927e5f9fe46afd8c342"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
