)]}'
{"id":"openstack%2Fpuppet-gnocchi~819966","triplet_id":"openstack%2Fpuppet-gnocchi~master~I8380b0ae90fc411ca1aee0ee4819d45cea888ac3","project":"openstack/puppet-gnocchi","branch":"master","topic":"prep-release-yoga-m2","attention_set":{},"removed_from_attention_set":{"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2022-01-07 14:34:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8380b0ae90fc411ca1aee0ee4819d45cea888ac3","subject":"Prepare Yoga M2","status":"MERGED","created":"2021-12-01 06:43:38.000000000","updated":"2022-01-07 14:34:02.000000000","submitted":"2022-01-07 14:34:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819966-prep-release-yoga-m2","meta_rev_id":"3773400084249b8d00abfe354a2089811b474fdd","_number":819966,"virtual_id_number":819966,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 14:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-01-07 14:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-01-07 14:34:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-01-07 14:19:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-01 06:54:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 15:35:04.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 14:19:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"9c1bc853ae431252f7416952158ba716c7497d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-01 06:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81cb85d65c00a3b7a9a52a3931561a5b546c78b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 06:54:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/768cbbb9aa274261af60c37f42ff01af : SUCCESS in 2m 24s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e10eea42d3c64558ba267ebe7bc7b889 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4aa516312b5ede73481703af8a66708ed44a4e57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-03 08:56:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bb772c4bb5503193c613459f4495fa00f74ca560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 09:06:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/a05ddd6367c64f7daecb0b897455a6f5 : SUCCESS in 2m 24s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39b5c18c584f460bad019ed99458abfb : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bd997af63c9264f6502668e20ef5139953ad7ff9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-13 15:35:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e8af83364407e204410b5f9c2b5c36c276ac2c18","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:40:14.000000000","message":"Topic prep-release-yoga-m1 removed","accounts_in_message":[],"_revision_number":2},{"id":"ab64172624556fd78ce18a2ee9c876b447f09eba","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:40:17.000000000","message":"Topic set to prep-release-yoga-m2","accounts_in_message":[],"_revision_number":2},{"id":"c9cdb82f950ee626db874e99397e38beb2327cdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:19:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4341b43da592c34782963e4a7c155d17192463ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:19:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c4c2f130bf2a17574933aa84ef09c88621a1baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:27:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/07f44e85231f497d856a2b8d5806e89f : SUCCESS in 3m 11s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12e2ab1e8d35499cb315af3d05dff05c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3bf1db3c4585a742a2888a472538b1cc85f2dca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:28:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c573cc6bc1fa65c6827b498d27a332bac45b1a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:34:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1e8dc0e469b34ab8ae6aa5c4ae2b298a : SUCCESS in 3m 21s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ea6120cc1f14aaa9356ebbb96572e1c : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"02e63e9617ce4360bfda3f4920027493b6efcec0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:34:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"09f36ae2183d7fa7e7c2ca036c207ffddd05c85d","revisions":{"a5bc504597583bb21bda999940588d2ff7de62b0":{"kind":"REWORK","_number":1,"created":"2021-12-01 06:43:38.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/66/819966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/66/819966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/1"}}},"commit":{"parents":[{"commit":"26b3401b20f75aaf43b4d7098767890abefd9ee6","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/26b3401b20f75aaf43b4d7098767890abefd9ee6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"subject":"Prepare Yoga M1","message":"Prepare Yoga M1\n\nUpdate the version metadata for Yoga milestone M1\n\nChange-Id: I8380b0ae90fc411ca1aee0ee4819d45cea888ac3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a5bc504597583bb21bda999940588d2ff7de62b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a5bc504597583bb21bda999940588d2ff7de62b0"}]},"branch":"refs/heads/master"},"68d708d00f24bcf23f7814e29d520f189573531d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-03 08:56:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/66/819966/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/66/819966/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/2"}}},"commit":{"parents":[{"commit":"26b3401b20f75aaf43b4d7098767890abefd9ee6","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/26b3401b20f75aaf43b4d7098767890abefd9ee6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-02 05:59:07.000000000","tz":480},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I8380b0ae90fc411ca1aee0ee4819d45cea888ac3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/68d708d00f24bcf23f7814e29d520f189573531d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/68d708d00f24bcf23f7814e29d520f189573531d"}]},"branch":"refs/heads/master"},"09f36ae2183d7fa7e7c2ca036c207ffddd05c85d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-07 14:19:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/819966/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/66/819966/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/66/819966/3"}}},"commit":{"parents":[{"commit":"701e66b4cae2703c7eefaa84d164d27fcca4a8f3","subject":"Load libraries in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/701e66b4cae2703c7eefaa84d164d27fcca4a8f3"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:52.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-07 14:19:19.000000000","tz":0},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I8380b0ae90fc411ca1aee0ee4819d45cea888ac3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/09f36ae2183d7fa7e7c2ca036c207ffddd05c85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/09f36ae2183d7fa7e7c2ca036c207ffddd05c85d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
