)]}'
{"id":"openstack%2Fpuppet-gnocchi~834624","triplet_id":"openstack%2Fpuppet-gnocchi~master~If01bfa3f2646e6a71f04e9b66d4fba8d68d540a9","project":"openstack/puppet-gnocchi","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-22 15:46:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If01bfa3f2646e6a71f04e9b66d4fba8d68d540a9","subject":"Use $::os_service_default instead of undef","status":"ABANDONED","created":"2022-03-22 07:56:05.000000000","updated":"2022-03-22 15:46:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dfd3c6193b24b65ed2a24cb5ed78e27637f58c7a","_number":834624,"virtual_id_number":834624,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-22 10:50:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 09:21:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a87c9887391f6706b16c84f0694eec6d6a6a246f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-22 07:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3395e73a07e6609307c058f0482caa48bf7bc4d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 09:21:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ec6527b763d47c6a3fa998a0322225e : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ec32a7a6d914cb9856cb437e5b6e146 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fc686150fa84f10a9f8655143a4d284 : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba40ea5aaa2f4995af54b0988c351a92 : FAILURE in 11m 58s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/855fdcca02c648279b66b98bbaad9144 : FAILURE in 11m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21eb6ab40fde47eea86cdbd450819bc4 : FAILURE in 9m 24s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f33f57c0a5b749e69e5afaedaebd1bdb : FAILURE in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0590e390beb4cf28e5915cfa04f1845 : FAILURE in 9m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e43baa456c641e79ccdb2c92f88afbd : FAILURE in 34m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a7ab6c69bfa4bc0bc1916512a9ef70c : SUCCESS in 1h 00m 33s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a572c15f5c614f698cce82d7fdb2df9c : SUCCESS in 48m 43s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bae9caf1a2748f6b6db9e0907febb95 : FAILURE in 38m 16s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e313a1725dcd4f7ba695cfe2b67cbe35 : SUCCESS in 1h 20m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7898e8e64fb8425f938225fc5d40fc24 : SUCCESS in 50m 11s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78618523d4c44a8fbb5ad6671aebe638 : SUCCESS in 14m 27s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d95f4915fa16454d98c79221a1fcc11f : SUCCESS in 20m 52s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/76d126581f8d48ec9ea59a4978b6f7b0 : FAILURE in 9m 21s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c17a970d33ad423ea9eb4055055828e5 : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a628e7abb30a4d038516a1d6ab7eefe8 : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db57a5ecfdb07df6dbebed740a53adaee569dce3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-22 09:23:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bad8c48d5e18934c0f67cda94ef86568e6a3cf1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 10:50:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc54b1aeb1964901b0f78f4176f46f11 : SUCCESS in 5m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6312d8a385354fc8bd2bd5b53064ddf4 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d940513428d4a97b102418109af2c95 : SUCCESS in 4m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f242fcdd2258462bbd2321678274c0e9 : SUCCESS in 12m 19s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/daecf9ee59204013ab541d43ca24f70d : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/378ede6b3a71487387eb0a2273f171b2 : SUCCESS in 9m 43s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4490086911a746c9bf705452a0fd3e8a : SUCCESS in 9m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd68e17663974777ad6d4a26bfb5837a : SUCCESS in 9m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1b5dd5fde8f44da80441bab6a619677 : FAILURE in 32m 19s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/04ca89a0d2754ad297e5d76dc8410ee8 : SUCCESS in 1h 00m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6617191281a46e6be59ecbf2fa51d3f : SUCCESS in 55m 54s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7720299a8fd4b399fc96b37b96f1528 : FAILURE in 38m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c94b0fa5a91f42a7a235d297a230ec86 : SUCCESS in 1h 12m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6c2e3c02ba24104afabc30598c09870 : SUCCESS in 1h 07m 46s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1832a8fe57df4da1931f8b622b87a4c8 : SUCCESS in 14m 58s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad3cfe0020834180abfa2e23e5bd7206 : SUCCESS in 20m 49s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b41930fc7f54e05ac8d1fad8b771de5 : FAILURE in 9m 20s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31b0e4b3388a49ce8c08b2fd40044e99 : SUCCESS in 22m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c43121f0289a49afb26c775baed0e952 : SUCCESS in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfd3c6193b24b65ed2a24cb5ed78e27637f58c7a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-22 15:46:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6bd1ae0831530b5280240d8dc14b1e14a06a34ca","revisions":{"924167a716024354e312f32f2da6409aecb88aea":{"kind":"REWORK","_number":1,"created":"2022-03-22 07:56:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/834624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/24/834624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/1"}}},"commit":{"parents":[{"commit":"d4416af16862af7bdb3914277814424fdbf857bb","subject":"Prepare Yoga RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d4416af16862af7bdb3914277814424fdbf857bb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 07:55:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 07:55:05.000000000","tz":540},"subject":"Use $::os_service_default instead of undef","message":"Use $::os_service_default instead of undef\n\n... because these two values are effectively (almost) same but\n$::os_service_default is now globally used as an indicator to use\nservice default values.\n\nChange-Id: If01bfa3f2646e6a71f04e9b66d4fba8d68d540a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/924167a716024354e312f32f2da6409aecb88aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/924167a716024354e312f32f2da6409aecb88aea"}]},"branch":"refs/heads/master"},"6bd1ae0831530b5280240d8dc14b1e14a06a34ca":{"kind":"REWORK","_number":2,"created":"2022-03-22 09:23:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/834624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/24/834624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/24/834624/2"}}},"commit":{"parents":[{"commit":"d4416af16862af7bdb3914277814424fdbf857bb","subject":"Prepare Yoga RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d4416af16862af7bdb3914277814424fdbf857bb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 07:55:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-22 09:23:42.000000000","tz":540},"subject":"Use $::os_service_default instead of undef","message":"Use $::os_service_default instead of undef\n\n... because these two values are effectively (almost) same but\n$::os_service_default is now globally used as an indicator to use\nservice default values.\n\nChange-Id: If01bfa3f2646e6a71f04e9b66d4fba8d68d540a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/6bd1ae0831530b5280240d8dc14b1e14a06a34ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/6bd1ae0831530b5280240d8dc14b1e14a06a34ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
