)]}'
{"id":"openstack%2Fpuppet-gnocchi~849173","triplet_id":"openstack%2Fpuppet-gnocchi~master~Iefc9b16662603fd8fc410c9080cb13b603bd4cf5","project":"openstack/puppet-gnocchi","branch":"master","topic":"bug/1981118","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-15 20:20:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iefc9b16662603fd8fc410c9080cb13b603bd4cf5","subject":"Switch provider to manage paste.ini file","status":"MERGED","created":"2022-07-09 04:10:40.000000000","updated":"2022-07-15 20:20:05.000000000","submitted":"2022-07-15 20:20:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849173-bug/1981118","meta_rev_id":"e015647e608db28103ab93a630f0973206238162","_number":849173,"virtual_id_number":849173,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 20:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-11 21:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-07-15 19:13:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-07-15 19:13:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 05:12:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 21:53:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-15 19:13:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3604c839edf0064fbe0d26c4d11ff1f224bbfb69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:10:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e440c7982dc48496f9c547f5869f853c1225ae3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 05:12:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4083c96d4a4bf2bdca1528049a8248\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51e58adf6467488da9b401b427f04ab0 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f289dcee11f14efa9b9538723b778383 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a79c4a6baad4ec78c5e6a62195636e7 : SUCCESS in 3m 02s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f097d61e9c4413a94390a1362e7938c : SUCCESS in 13m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09074bcfc146405e833506b11d8a2401 : SUCCESS in 9m 23s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60eeacb515ac434188096ac584a04c4a : SUCCESS in 7m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e158e98c6cf64593ae125fa3043cc99a : SUCCESS in 10m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1533da39387646edaa728e22ad6fc18d : FAILURE in 23m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d6add0291e44bc6a319505e885e0a62 : SUCCESS in 54m 43s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4091a62d4bd545d482878bbfbff75c06 : FAILURE in 42m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/759ffcead97f4a11995e919bd0eb9be4 : SUCCESS in 1h 00m 22s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25aca270efee41dea58acc098c4056d9 : SUCCESS in 12m 50s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/de9c68af715d4d0c933e6b46d5fd3329 : SUCCESS in 20m 40s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aac9bde10b045ecbe3de394749854f5 : SUCCESS in 15m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd4ae3bd99fb437a88b5f2ad9fd57ddf : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1639171bbf6d9b17efa47d5abd5018bf1844abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:05:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8ec4e5cb8d90f5832845866f930770c7f71bac23","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:05:08.000000000","message":"Topic paste-ini-provider removed","accounts_in_message":[],"_revision_number":2},{"id":"c89a37e8069fa2f7f3a1f00d991037690121935a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:05:12.000000000","message":"Topic set to bug/1981118","accounts_in_message":[],"_revision_number":2},{"id":"2a62679791fbbb2be251e9b76e29ac3624236d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 11:03:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9281d603ab4924bff892c91c931f18\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60e3db60fe0447de8ee0e5f38d5f1ac8 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c348a051fede4277852b5ba9d00c742b : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aa34ccdf8d74ec7938d180d432624f1 : SUCCESS in 2m 55s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ab40daf649841c3a06287a5a2fef4eb : SUCCESS in 10m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f85e87515cd433382e3300f71214df7 : SUCCESS in 11m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d6b330d0d2b4e06a28d7a1e5bbb4e01 : SUCCESS in 6m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc112b063ff746cfb459121135b07ca7 : SUCCESS in 5m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efffa86ce1164b05938068e6ffc00b30 : FAILURE in 26m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c2232224c7040229f7c5f0307b45255 : SUCCESS in 50m 35s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98a3ae3c20244c62bd74e07b6a3def08 : FAILURE in 41m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea2365864bd94589b65c211183980ebb : SUCCESS in 56m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0bf31dd81647e89a058b2a0f30b38d : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/7842ccc116464dc0ac84610ffae88947 : SUCCESS in 21m 04s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1290858681db44e795b963bcac66b90a : SUCCESS in 13m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60d2caea7ec84729821ec4bb99ae8fe8 : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e6cb4ec28b8e056ec0262722322eed79da902cf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 21:53:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"90cbc61d253d18bf257cf80486412cb3a992d5ed","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 19:13:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb244245a9695dd0481ed300b68d270328664110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 19:15:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61857a11047fae0406861097b4e3768222d3ff3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:20:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4e893a3f574b128776537e9d9b03f3\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fa58f06a3bb4a5fabbc6b8a8e22d72a : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcfb6fe9e8cc405d9608b38dfdf06aa2 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bfc50339ec246eb94d9f2cc8bc1c0f6 : SUCCESS in 3m 23s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/625099ba85b740319d414a8888f5e375 : SUCCESS in 14m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc80166bc0514ac18a0aaf4d52c446b2 : SUCCESS in 10m 09s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f8621400354452483fee461e79aaae2 : SUCCESS in 10m 02s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5793b65afa254f7c9ae3aeb679b9cddc : SUCCESS in 54m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9731ebe9578344edb5435eb18a30ca7e : SUCCESS in 1h 04m 03s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9db30c1636044476aeff28c2677b6720 : SUCCESS in 20m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e015647e608db28103ab93a630f0973206238162","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:20:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ef5d0de458bb2f0aaeb97755f0c26aa3d85d814","revisions":{"9bb3cf4abb04ead989a5651ec569febea6667f5f":{"kind":"REWORK","_number":1,"created":"2022-07-09 04:10:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/849173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/73/849173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/1"}}},"commit":{"parents":[{"commit":"b2516674df7e3363b527bf34478f4dcb932c2e58","subject":"Fix typos in parameter descriptions and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b2516674df7e3363b527bf34478f4dcb932c2e58"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:10:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:10:24.000000000","tz":540},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe resource types to manage the \u003cservice\u003e.conf files.\n\nChange-Id: Iefc9b16662603fd8fc410c9080cb13b603bd4cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/9bb3cf4abb04ead989a5651ec569febea6667f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/9bb3cf4abb04ead989a5651ec569febea6667f5f"}]},"branch":"refs/heads/master"},"4ef5d0de458bb2f0aaeb97755f0c26aa3d85d814":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-09 10:05:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/849173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/73/849173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/73/849173/2"}}},"commit":{"parents":[{"commit":"b2516674df7e3363b527bf34478f4dcb932c2e58","subject":"Fix typos in parameter descriptions and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b2516674df7e3363b527bf34478f4dcb932c2e58"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:10:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:05:00.000000000","tz":0},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe resource types to manage the \u003cservice\u003e.conf files.\n\nCloses-Bug: #1981118\nChange-Id: Iefc9b16662603fd8fc410c9080cb13b603bd4cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/4ef5d0de458bb2f0aaeb97755f0c26aa3d85d814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/4ef5d0de458bb2f0aaeb97755f0c26aa3d85d814"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
