)]}'
{"id":"openstack%2Fpuppet-gnocchi~854740","triplet_id":"openstack%2Fpuppet-gnocchi~master~Ifa33b2e84284d8032c4e4f54acb26c30fc285cb5","project":"openstack/puppet-gnocchi","branch":"master","topic":"apache-logging","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-26 17:18:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa33b2e84284d8032c4e4f54acb26c30fc285cb5","subject":"Add Apache WSGI logging parameters for pipe/syslog","status":"MERGED","created":"2022-08-26 06:42:11.000000000","updated":"2022-08-26 17:19:34.000000000","submitted":"2022-08-26 17:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854740-apache-logging","meta_rev_id":"51384d8011ef742ec5e4120ce1781894033922b0","_number":854740,"virtual_id_number":854740,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 17:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-26 11:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-26 15:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-26 15:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 08:10:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 11:18:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-26 15:58:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c46fffd72dfe61a2f5e7b3aee02ffa913fc95a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 06:42:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d28d748b1a669742c213b07b65b823b4a1a20733","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 06:42:56.000000000","message":"Topic set to apache-logging","accounts_in_message":[],"_revision_number":1},{"id":"5364decf00410efa3add6bd364bafc9acbd1e6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:10:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69ff4a0275a7492a9bb037280a45f633\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8dea7c5ec9447a387f6fd0b21059338 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed47f2e3b96445eba4ed480625a5bdb2 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/300f0c3a3e434f7fa395ca4ced62ecf4 : SUCCESS in 4m 17s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9cbb2fab6e34e3184f5678ae5228cae : SUCCESS in 16m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ee8b0f9c44c4c9dac8a8fd62f312571 : SUCCESS in 8m 18s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a5188bb09342cb8acb2a9ed3bcaa29 : SUCCESS in 11m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbe1d022ef5445049f3c10f07bdb9f98 : SUCCESS in 10m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/529904aaabf1461bae3e0b225c55659c : FAILURE in 48m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0314bafc43a54315a59025a453ca12d9 : SUCCESS in 58m 22s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74b04895e71b465d8988e74347e45868 : FAILURE in 56m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/24ebfcfb07ec4553ac65e160bfe52ac4 : SUCCESS in 1h 22m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/199c744532074df79507258db3eb8457 : SUCCESS in 18m 03s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a53a9900c8e42e791f68325ae9b5d73 : SUCCESS in 22m 52s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c87f1953a73e467da74f1eea4dc5a615 : SUCCESS in 18m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e21bc6cfc054c8fbaaccc8c01a942b1 : SUCCESS in 13m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7243f2b597834cbe83e9d9956c86fa26 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8249580c0de92fd0130a8882cf1ed0a91537b75a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 08:41:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ecdd1c670f51945df57a5bb3844ffa3dd141348e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 09:55:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f624af145f74dc8b92c08e71e90bd7e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82b5e2d051024064b20fc3d092981212 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a2791bfa6ef40d2848ce887eeaa5494 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ce5e78689be4cafb76a0c4e888397d8 : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/07e8db0738084bfbaf0c255bdc79bb15 : SUCCESS in 15m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cc8082386f84483a1c36477fc4d38c6 : SUCCESS in 11m 03s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee3c8e14e78043aa848b1350fdc6c093 : SUCCESS in 11m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d80a96ee85614a01a185509b824a0ec8 : SUCCESS in 11m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17bb7e696514df8bcb4dfdfcf1eaa0c : FAILURE in 48m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49368714889743d69c2166425819ee94 : SUCCESS in 1h 07m 23s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de402e9de8c74507a875641ba68bf33e : FAILURE in 1h 00m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/937cc10c2f5c4e6696ffa2cdd89c9429 : SUCCESS in 1h 02m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea023c64e1a14ce4b638c48e7e6501c3 : SUCCESS in 18m 17s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9262674a5fe4f96b0e772f0e033a8e8 : SUCCESS in 30m 36s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe360de046e649d7904846b7e9a84418 : SUCCESS in 15m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/997c1bef16a9450da9ce1fa7c8e722cf : SUCCESS in 17m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b20a2ab858634f65abccab673e62ce1c : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"113beb5ad0f153b2c6eb09dcdbef0e27b373b5f1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-26 11:18:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ec53fd10384551c4983831f6f6e9dd98058eb08e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-26 15:58:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1281d15101afbc3696ee99dfe51bcdb33f7fab8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 15:59:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ad5302a9d5fd0ce05ecc156f5f45523853238d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:18:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4de9f15d86bb4caaaec0fe639eb17ff2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8be55e5f64b43029981a7811f9fa553 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3a15edb19da46748da5ceeaf6650e5c : SUCCESS in 3m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77056c18c2bb41368e9f5c54f0a38ead : SUCCESS in 3m 49s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3abe13f03cac4d8d8085ad9fbef22060 : SUCCESS in 17m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c201de30f2914128b36add84be0c53aa : SUCCESS in 11m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/528b498aeac94781a788cf10a9d6cbd7 : SUCCESS in 11m 29s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d9568e3ad8b47c4a69ceca3b82b99e4 : SUCCESS in 1h 02m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff89d2afb19c40bda96632c896b17e1d : SUCCESS in 1h 17m 13s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/77695691d092455180fb4958376b04c1 : SUCCESS in 22m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05a8f9dc3193412fa1039cc8845ba922 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"6b1c47d4dca624e95d86b0b7e4e8af8e3c3b74bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:18:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"51384d8011ef742ec5e4120ce1781894033922b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 17:19:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4960fb08a9d42e5ba7bedafef7c95da\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5940b06060e04592af45fbadea5a7258 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"990298ddd51b9502b11f875ac8b5ccdab5fbcc8e","revisions":{"6996eef838072a38638f0136b66a0512d7dfc07a":{"kind":"REWORK","_number":1,"created":"2022-08-26 06:42:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/854740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/40/854740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/1"}}},"commit":{"parents":[{"commit":"8001209a81ca78bea86933b86fa96bc689477957","subject":"Remove gnocchi::metricd::cleanup_delay","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/8001209a81ca78bea86933b86fa96bc689477957"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 06:41:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 06:42:01.000000000","tz":540},"subject":"Add Apache WSGI logging parameters for pipe/syslog","message":"Add Apache WSGI logging parameters for pipe/syslog\n\nAdd parameters for advanced logging configurations in Apache to\nsupport piped logging and support for syslog (via mod_syslog\navailable in Apache \u003e\u003d 2.5.0)\n\nCo-Authored-By: Andy Botting \u003candy@andybotting.com\u003e\nChange-Id: Ifa33b2e84284d8032c4e4f54acb26c30fc285cb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/6996eef838072a38638f0136b66a0512d7dfc07a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/6996eef838072a38638f0136b66a0512d7dfc07a"}]},"branch":"refs/heads/master"},"990298ddd51b9502b11f875ac8b5ccdab5fbcc8e":{"kind":"REWORK","_number":2,"created":"2022-08-26 08:41:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/854740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/40/854740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/40/854740/2"}}},"commit":{"parents":[{"commit":"8001209a81ca78bea86933b86fa96bc689477957","subject":"Remove gnocchi::metricd::cleanup_delay","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/8001209a81ca78bea86933b86fa96bc689477957"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 06:41:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 08:41:09.000000000","tz":540},"subject":"Add Apache WSGI logging parameters for pipe/syslog","message":"Add Apache WSGI logging parameters for pipe/syslog\n\nAdd parameters for advanced logging configurations in Apache to\nsupport piped logging and support for syslog (via mod_syslog\navailable in Apache \u003e\u003d 2.5.0)\n\nCo-Authored-By: Andy Botting \u003candy@andybotting.com\u003e\nChange-Id: Ifa33b2e84284d8032c4e4f54acb26c30fc285cb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/990298ddd51b9502b11f875ac8b5ccdab5fbcc8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/990298ddd51b9502b11f875ac8b5ccdab5fbcc8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
