)]}'
{"id":"openstack%2Fpuppet-gnocchi~862438","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Fwallaby~I2deed38f61aebb0617a55f12070b46cb0c20bef0","project":"openstack/puppet-gnocchi","branch":"stable/wallaby","topic":"puppet-wallaby-em","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-31 13:06:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I2deed38f61aebb0617a55f12070b46cb0c20bef0","subject":"Prepare the final wallaby release","status":"MERGED","created":"2022-10-24 00:41:03.000000000","updated":"2022-10-31 13:22:11.000000000","submitted":"2022-10-31 13:22:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862438-puppet-wallaby-em","meta_rev_id":"750cd9686091f26b685002886894dff18d9170fe","_number":862438,"virtual_id_number":862438,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-31 13:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-10-24 15:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-10-30 20:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-10-30 20:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-24 00:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 15:43:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-10-30 20:38:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7c7d31eb7e6b6243e74e45ae9419a4189499dd04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-24 00:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d64c8015096c824be67b3a8951f5269290efec4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-24 00:41:32.000000000","message":"Topic set to puppet-wallaby-em","accounts_in_message":[],"_revision_number":1},{"id":"dfe0bdb626b1ae830b701da4e9367b1ab36791a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 00:43:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9de64c8e666d41ba96a1564102867481\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/286539fb3225463a8b73092aba503dc8 : SUCCESS in 1m 52s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c10eeaf051346d897f9510796602fbe : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4b8fa11d409617f169494c291d8e72199f968d8d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-10-24 15:43:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b41920e5a45ec7f05fcde81e9ffd90adf840938a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-30 20:38:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0107dadf94fa5b15c3fe3f16d89c8920805dd15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 20:38:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f8460ecbbbf9b5006ddacfdfe7c254713974418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 20:42:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c48738a30084eb98a98cd4df7880c8d\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/cbf2bffeaa2f46be9405b0ba14da0a94 : FAILURE in 1m 48s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12e62427bed94520a4922421340bf7a3 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"68c8bf063a2e03ca99956555e50473701385d837","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-31 13:06:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb7b2eeb956d38af1a9b665aebe577b85b0f3d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 13:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb0f7802acb40bfaa86af1aa23335a5\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/751f672ce1de4c0e85bc8a8441595911 : SUCCESS in 2m 05s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dea6df5ceceb4f5a82c4ad9374d593d4 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b316a7d2466397199a37b61ff37c7e1e018e6718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 13:15:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79d9d3d1abe82fa93dbcc0662a5713dde5b6d4c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 13:22:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d91071488ea47f699df0885a8d3f727\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/46584808374447acae2cbfa3eb814501 : SUCCESS in 1m 58s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e55eafbf7dd47bc939a88bd8d157bbf : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"750cd9686091f26b685002886894dff18d9170fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 13:22:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b93585956d95423fb05d30f82318048df15bda5b","revisions":{"b93585956d95423fb05d30f82318048df15bda5b":{"kind":"REWORK","_number":1,"created":"2022-10-24 00:41:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/862438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/38/862438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/862438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/862438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/862438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/862438/1"}}},"commit":{"parents":[{"commit":"72717d6e545bdf2a216bd23cebab8ed3397e5271","subject":"s3_access_key_id should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/72717d6e545bdf2a216bd23cebab8ed3397e5271"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-24 00:40:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-24 00:41:00.000000000","tz":540},"subject":"Prepare the final wallaby release","message":"Prepare the final wallaby release\n\n... to create a release with one bug fix[1], which was merged after\nthe last wallaby release.\n\n[1] 72717d6e545bdf2a216bd23cebab8ed3397e5271\n\nChange-Id: I2deed38f61aebb0617a55f12070b46cb0c20bef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b93585956d95423fb05d30f82318048df15bda5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b93585956d95423fb05d30f82318048df15bda5b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
