)]}'
{"id":"openstack%2Fpuppet-gnocchi~870441","triplet_id":"openstack%2Fpuppet-gnocchi~master~I46de72deb389f6f0c4b0af2f9f17416334f5f891","project":"openstack/puppet-gnocchi","branch":"master","topic":"puppet-jammy","hashtags":[],"change_id":"I46de72deb389f6f0c4b0af2f9f17416334f5f891","subject":"Switch to Ubuntu Jammy (22.04)","status":"MERGED","created":"2023-01-15 13:21:17.000000000","updated":"2023-01-16 00:36:25.000000000","submitted":"2023-01-16 00:35:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870441-puppet-jammy","meta_rev_id":"e23bbdbfdee619786571ee4f6ee8f22c3b0929bd","_number":870441,"virtual_id_number":870441,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 00:35:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-01-16 00:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-01-16 00:22:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-15 13:33:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84c1c50b63898440e8855f411a6154d05fe6a034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-15 13:21:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdc9690329c09af35227a0c2df5f24af16cf818b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-15 13:33:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/283827b5d1f449c4a549fd098d677f0e\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ea656929ef7741aeb8cf4fa7b3aef422 : SUCCESS in 2m 31s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94f58cf540964f3a92d7f96e917bb2af : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f1ffb3e7e304c7cba7131e161f718a1 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"93f323a5df3487b50d092696185c479618b3f500","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-16 00:22:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f0501070cba6645b57f7b85e842c793c62f9c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:22:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d59e9efe83efc5452df089779f3c940c1403813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:35:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e14d91204b54467689267d3f1b3089ce\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b0aba6fe46b04d5abcc018da499ee82a : SUCCESS in 1m 47s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a180fd41184706a29825edd5a3fa8f : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2075f3d1bb1a4c60a8cc69aa0c4c439c : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"76e2b9e5883ac5acd1b3ee034b1ab8d265116e4d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:35:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e23bbdbfdee619786571ee4f6ee8f22c3b0929bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:36:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5cb214aeb174bebb2e7e62598fd33ee\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d767a494bac64144a47857ac6cf677f6 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa6ef57180a7876e077f019e6befa28d961a2456","revisions":{"fa6ef57180a7876e077f019e6befa28d961a2456":{"kind":"REWORK","_number":1,"created":"2023-01-15 13:21:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/870441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/41/870441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/41/870441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/41/870441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/41/870441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/41/870441/1"}}},"commit":{"parents":[{"commit":"aaceae30fbd07bfaeb1dff1d39452b1244e5802f","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/aaceae30fbd07bfaeb1dff1d39452b1244e5802f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-15 13:21:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-15 13:21:05.000000000","tz":540},"subject":"Switch to Ubuntu Jammy (22.04)","message":"Switch to Ubuntu Jammy (22.04)\n\n... because Focal no longer supports the recent releases such as Zed.\n\nChange-Id: I46de72deb389f6f0c4b0af2f9f17416334f5f891\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/fa6ef57180a7876e077f019e6befa28d961a2456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/fa6ef57180a7876e077f019e6befa28d961a2456"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
