)]}'
{"id":"openstack%2Fpuppet-gnocchi~871550","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Fwallaby~I6a99bd57ad8e3713e824e4394b6fc2125167a768","project":"openstack/puppet-gnocchi","branch":"stable/wallaby","topic":"multistropt-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-31 09:31:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a99bd57ad8e3713e824e4394b6fc2125167a768","subject":"Add support for MultiStrOpt","status":"MERGED","created":"2023-01-24 01:17:14.000000000","updated":"2023-01-31 09:31:44.000000000","submitted":"2023-01-31 09:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871550-multistropt-stable/wallaby","cherry_pick_of_change":800605,"cherry_pick_of_patch_set":1,"meta_rev_id":"da6fa1833dec50f3cbccc91298ea44bd741eac11","_number":871550,"virtual_id_number":871550,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 09:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-01-26 17:48:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-01-31 08:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-01-31 08:13:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-24 02:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 17:48:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-01-31 08:13:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"bfffb4999d6d5183006cb12555449bb70b264c36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-24 01:17:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"29863d20bbfa85cfb195763df67eacc837f73945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 02:26:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53e396f548e84d02b7bd25f941ebd30f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e38dba6326514ba7a522fb314ed0529c : FAILURE in 2m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb2c7341b821467fae239d46a894de97 : FAILURE in 2m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5773cf7c272948b18e25c3e30d81c0d1 : FAILURE in 8m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96293d5642a845f6b390ac1a97c31248 : FAILURE in 5m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c8df0b8af754ca5bce9c8a8cf4f39ba : FAILURE in 5m 14s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/506b8db554364baf8d1a1c409e71ce46 : FAILURE in 28m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/323f27f911a949f0963af7d7c5b5e9a6 : SUCCESS in 1h 04m 38s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/71ae48da6bea4ad59b3524e323f0b8ab : FAILURE in 13m 56s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58d1e44a99f244a7b59574f488cbdce5 : FAILURE in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55d99269901bef6b8e8d71fec4aa84dd2b86ebeb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-25 05:25:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1f54f4855f10d7a9d0db498583f069e6b9b8fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 06:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba193c88dee44f9885efddad80bf967d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91326283fe9640f2b06f6efe9d066c18 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/768114389dca48988c4d6eed893ae590 : SUCCESS in 2m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5186f31bd7b04898962f832c6091f0d7 : SUCCESS in 11m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c5f80db3a574053a10d6494828e392f : SUCCESS in 7m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c2d0b5459334777b0b4be51076fdaa7 : SUCCESS in 7m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31338cf0a1c244e4872e2c323f18d5ff : FAILURE in 39m 20s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c75de95a24f0440591517b303da527cd : SUCCESS in 1h 09m 20s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/47a6b6c615f941618d424cd6a32b7c6d : SUCCESS in 20m 28s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1601f2f1b1314528946c3dc645d381ee : FAILURE in 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcb00073bbbe93ee04b0835a1147f82f31a9c240","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-26 17:48:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e116d9de19304e7fe19adf168856f17ff9fe5c6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-31 08:13:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"daf4324b5b275b240bb07005a8abc3b336755dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 08:14:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0af120c94123f5b6421293544b55f4511bb81112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 09:31:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/039fc706a3af417fbd8396eb4f103bd7\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfd4845dad6241d3aa42c9578ae5d298 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d26a0aae9b5483e9625c255d3fe3e87 : SUCCESS in 2m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7033612c4dc4ea1b6b60d98f5c36ed1 : SUCCESS in 11m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e45d6b46aa94d62a308e1af2729f8be : SUCCESS in 6m 06s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a853b039dad243608d4421b71f095acb : SUCCESS in 1h 12m 21s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/08465e6d9ad4438ebe92cbbdb56d1bdf : SUCCESS in 21m 45s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d653b8047534015a7ccce3690b951b8 : FAILURE in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da6fa1833dec50f3cbccc91298ea44bd741eac11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 09:31:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7584b9495ac09e94f4c3406227ab8f2931e1e318","revisions":{"7584b9495ac09e94f4c3406227ab8f2931e1e318":{"kind":"REWORK","_number":1,"created":"2023-01-24 01:17:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/871550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/50/871550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/50/871550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/50/871550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/50/871550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/50/871550/1"}}},"commit":{"parents":[{"commit":"457dee3856445b624e2de8be34e3602c6b270016","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/457dee3856445b624e2de8be34e3602c6b270016"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-13 09:02:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-24 01:17:14.000000000","tz":0},"subject":"Add support for MultiStrOpt","message":"Add support for MultiStrOpt\n\nThis replaces the provider implementation of gnocchi_config type so\nthat MultiStrOpt, which is used by several options like\n - oslo_policy/policy_dirs\nare handled correctly.\n\nChange-Id: I6a99bd57ad8e3713e824e4394b6fc2125167a768\n(cherry picked from commit f41c7d8e78d42caee68c709ac9bd3996cbfb3acd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7584b9495ac09e94f4c3406227ab8f2931e1e318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7584b9495ac09e94f4c3406227ab8f2931e1e318"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
