)]}'
{"id":"openstack%2Fpuppet-gnocchi~900268","triplet_id":"openstack%2Fpuppet-gnocchi~master~I891f1eb7cacfbf016c8d57dd65f47ab4fed7394c","project":"openstack/puppet-gnocchi","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-09-16 14:29:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I891f1eb7cacfbf016c8d57dd65f47ab4fed7394c","subject":"Make it possible to customize indexer/url","status":"ABANDONED","created":"2023-11-07 08:13:18.000000000","updated":"2024-09-16 14:29:39.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f8531e59ab70bb1f4cb59cf8729f04a346f0ef33","_number":900268,"virtual_id_number":900268,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-07 09:39:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-11-07 09:47:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 09:39:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 09:47:21.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"b0f08130cebdaf7417c1eb7838b3282c3473b7eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-11-07 08:13:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"352fbd6a39e589a3e03d3518517b24a6bcd925e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 09:39:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1769a37fb02b431b812bacecbbfad3a4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30edf7361e6f477a81a4044154e2a1a1 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92d0d507942e420b9c619eb6ba5fd98c : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/810ddc2ad68b436eb9617d55aef6d096 : SUCCESS in 5m 02s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/9659ed43c398445a8cd90524b1168dac : SUCCESS in 11m 10s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ed96bf4338f450288b4d6cb40ccd4be : SUCCESS in 6m 54s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/3934a38276ee4120a4695fdfc779702b : SUCCESS in 10m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/069ee5ba2a4a4255bdd7ccce45251aa8 : SUCCESS in 6m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d1b233c3fb645858b0f1258b2651fbd : SUCCESS in 40m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/edc1259cb1cf477d90d8c35a6ae0bdd5 : SUCCESS in 1h 01m 56s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e83e3b900724846ab9499eca011d8f2 : SUCCESS in 1h 00m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/701e331cbd99474dab145d830f2890d6 : SUCCESS in 1h 18m 00s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/c44e47d9743247ffb2a2c804c8d57dad : SUCCESS in 21m 40s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8144833575c4374b85e39d4db6e4ac7 : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/161f66f569b244478feea2e177e2ae0d : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4284ab86d04d7eaf3273cc0e9f97b9b8992510f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-07 09:47:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8531e59ab70bb1f4cb59cf8729f04a346f0ef33","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-16 14:29:39.000000000","message":"Abandoned\n\nI\u0027ll abandon this because of no feedback to the given -1 . Please feel free to reopen this if you think the problem is valid.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1094e52faf4951703feea22dce2253679d38d8e","revisions":{"a1094e52faf4951703feea22dce2253679d38d8e":{"kind":"REWORK","_number":1,"created":"2023-11-07 08:13:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/68/900268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/68/900268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/68/900268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/68/900268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/68/900268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/68/900268/1"}}},"commit":{"parents":[{"commit":"0243f0eebf9e2612a28d95c33ea9dd4cd81a48d9","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0243f0eebf9e2612a28d95c33ea9dd4cd81a48d9"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-11-07 08:11:06.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-11-07 08:11:06.000000000","tz":60},"subject":"Make it possible to customize indexer/url","message":"Make it possible to customize indexer/url\n\nIt used to be that indexer/url was always taking the same value as\nfor the database_connection. However, in some case, it may be useful\nto make it different. In our case, we want database_connection to\naccess a galera node with read/write access, while the [indexer]/url\ncan go to a slave node, in our case, localhost. This helps us a lot\nto spread the load in our galera cluster (as our billing SQL is like\nsuper busy).\n\nChange-Id: I891f1eb7cacfbf016c8d57dd65f47ab4fed7394c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a1094e52faf4951703feea22dce2253679d38d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a1094e52faf4951703feea22dce2253679d38d8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
