)]}'
{"id":"openstack%2Fpuppet-gnocchi~933546","triplet_id":"openstack%2Fpuppet-gnocchi~master~I248a7ab2287c8252478b25f020fc5e5f5e653813","project":"openstack/puppet-gnocchi","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-29 17:45:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I248a7ab2287c8252478b25f020fc5e5f5e653813","subject":"Support customizing oslo.db options","status":"MERGED","created":"2024-10-27 17:31:44.000000000","updated":"2024-10-29 17:46:54.000000000","submitted":"2024-10-29 17:45:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933546","meta_rev_id":"46e35f9c50fecedeb5c2452050637852ecdf0f97","_number":933546,"virtual_id_number":933546,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 17:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-10-28 07:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-10-29 16:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-29 16:19:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-27 18:41:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-28 07:11:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-29 16:19:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"04fb8d4b5bdeff1de122b6f62094c90ca8ce2050","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-27 17:31:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58d75e16d9acff70411a169ac2771f5cb9edd774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-27 17:32:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85bd95d9226dbd914f5764911d2eaba5fee40df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-27 17:37:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"18ffaf2b2714f4ebaab68919e9c9d841b22dfba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 18:41:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cf9e12d6c1b4aaebc291d37b8a15b6f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55e3b0e261cb4aea9e8d9537e3b05f37 : SUCCESS in 2m 18s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01466c65bd2e429d893682309ead8266 : SUCCESS in 2m 23s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8301f8cdace34389b0cd72afbb6ee915 : SUCCESS in 5m 14s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e702d1ac2a6c426b9293d4b64e4b6859 : FAILURE in 8m 55s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9010ff465f81413aa2023bad9f9859e6 : FAILURE in 5m 12s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f133b56a44b40d08fe20e72c0d8b3c0 : FAILURE in 5m 02s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae7e65998f3f4543b014ec2e7158fab0 : FAILURE in 5m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0788f1e9df1439b890b776d1325be87 : SUCCESS in 47m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/627e7dc16a1a496c9f5547b036d365b7 : SUCCESS in 34m 19s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92c7fb2ea4aa4b96b1a1016c2619b1a7 : SUCCESS in 55m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d0759a06f434ab5b7362ebc3945c7b8 : SUCCESS in 45m 08s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cc69177187c423bbb7850d1267ca795 : SUCCESS in 12m 54s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92195c6de7714ba19e9fd65ce623da77 : SUCCESS in 10m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae3cad66b80141bdb9eedce98b8cb981 : SUCCESS in 8m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67f581db1f994fee92fd0bbc6780796f : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"a3f2e64a5f3454bf007ea362a999a6b991c5c018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-27 19:03:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"be0de2339957a3bc2222b4c19411e0bba7fe995a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 20:14:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5ea4e621be4c53b8a0e2607cce9e68\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0d6e241803644569eb2b88d379c01b1 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7062f3adcd2d413296db9073ed317909 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3a71c9b2dc44f9d8dfe10453b40b9e9 : SUCCESS in 4m 18s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/5892e87fbfdf41418d71a7ebcadb0a36 : SUCCESS in 7m 51s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a023601ff4c240589557e2ac15c25595 : SUCCESS in 4m 31s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b02ddf5c26c4ae9b5b7b13fae560881 : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a00fc7b30ef5446692a0cfb0493f2218 : SUCCESS in 4m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b50bc9b5787d4aeaa707409138fb901f : SUCCESS in 33m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/551f28212bb54492a3ed398d275f1a60 : SUCCESS in 52m 10s\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f64482d02d5a4a1282571e6bf78b6c3b : SUCCESS in 44m 46s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/926dd90267cd447cad4b950b45077307 : SUCCESS in 1h 07m 59s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad8b4dc5abf7437dbf9e5e04c633b335 : SUCCESS in 11m 46s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e78b689d17d4bca8c4685cfd811c289 : SUCCESS in 9m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28bb150ec69b4bea8d5585e2af97a0c8 : SUCCESS in 12m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ff7dcec944745a8956c3dcaed7412b0 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":4},{"id":"d41789736c7a466f50af569dcde389cbbcb36b39","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-28 07:11:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b8bfad03cc1d489e39a285b99946789cf2519884","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-29 16:19:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4322cc9f8c13a4538d0fa5e4a1e8d2ccefdab6eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 16:21:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4a0ee67b998af3d4dd0a6c1272c5521488e42327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:45:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c338441dafe4a2886e39edb1a1c14dd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dd9950bd51d4facab70188b46e6fe15 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7383162f5f224b9292ea8e9c9ff4179d : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9e865b05d564371a0f8579c503479be : SUCCESS in 9m 13s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a422a5afaf344fcaf4908dcce8fb2a0 : SUCCESS in 5m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04048937f0a244af856994a71bfdc3b5 : SUCCESS in 1h 02m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/305cf2042f2b46679065c2dcd18d4350 : SUCCESS in 1h 18m 27s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb52215d8d54427fa32227e557700103 : SUCCESS in 18m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bd1d488e4c947de9048eb52a5c98891 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":4},{"id":"6cda81c730151c6b42cd9c49ac07170446f34214","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:45:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"46e35f9c50fecedeb5c2452050637852ecdf0f97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 17:46:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1593490f4d7f48c5be6aaa6e7b86d089\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bee82ad28424c27b8e9726709131aa2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"67d29a9f29c55d2be4341a1af546ee9599b3d429","revisions":{"868bfe4f9fb17508341ad6d5f24af1aad1b9dda1":{"kind":"REWORK","_number":1,"created":"2024-10-27 17:31:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/933546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/46/933546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/1"}}},"commit":{"parents":[{"commit":"7929bd422d02c45bfa9e334a3f52739f3d8ff9ef","subject":"Merge \"Expose configure_service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7929bd422d02c45bfa9e334a3f52739f3d8ff9ef"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 17:28:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 17:28:56.000000000","tz":540},"subject":"Support customizing oslo.db options","message":"Support customizing oslo.db options\n\nGnocchi uses oslo.db internally when mysql or PostgreSQL is used as\nthe indexer backend. Expose oslo.db options so that users can customize\nsome of these options for tunings.\n\nChange-Id: I248a7ab2287c8252478b25f020fc5e5f5e653813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/868bfe4f9fb17508341ad6d5f24af1aad1b9dda1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/868bfe4f9fb17508341ad6d5f24af1aad1b9dda1"}]},"branch":"refs/heads/master"},"c352a384c6cc3633d57e1e6a3f0a9f0b85d05f4a":{"kind":"REWORK","_number":2,"created":"2024-10-27 17:32:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/933546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/46/933546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/2"}}},"commit":{"parents":[{"commit":"7929bd422d02c45bfa9e334a3f52739f3d8ff9ef","subject":"Merge \"Expose configure_service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7929bd422d02c45bfa9e334a3f52739f3d8ff9ef"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 17:28:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 17:32:33.000000000","tz":540},"subject":"Support customizing oslo.db options","message":"Support customizing oslo.db options\n\nGnocchi uses oslo.db internally when mysql or PostgreSQL is used as\nthe indexer backend. Expose oslo.db options so that users can customize\nsome of these options for tunings.\n\nChange-Id: I248a7ab2287c8252478b25f020fc5e5f5e653813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/c352a384c6cc3633d57e1e6a3f0a9f0b85d05f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/c352a384c6cc3633d57e1e6a3f0a9f0b85d05f4a"}]},"branch":"refs/heads/master"},"85405c2cdde60879ba2e1d70042513bbe6ca3d6e":{"kind":"REWORK","_number":3,"created":"2024-10-27 17:37:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/933546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/46/933546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/3"}}},"commit":{"parents":[{"commit":"7929bd422d02c45bfa9e334a3f52739f3d8ff9ef","subject":"Merge \"Expose configure_service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7929bd422d02c45bfa9e334a3f52739f3d8ff9ef"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 17:28:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 17:37:31.000000000","tz":540},"subject":"Support customizing oslo.db options","message":"Support customizing oslo.db options\n\nGnocchi uses oslo.db internally when mysql or PostgreSQL is used as\nthe indexer backend. Expose oslo.db options so that users can customize\nsome of these options for tunings.\n\nChange-Id: I248a7ab2287c8252478b25f020fc5e5f5e653813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/85405c2cdde60879ba2e1d70042513bbe6ca3d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/85405c2cdde60879ba2e1d70042513bbe6ca3d6e"}]},"branch":"refs/heads/master"},"67d29a9f29c55d2be4341a1af546ee9599b3d429":{"kind":"REWORK","_number":4,"created":"2024-10-27 19:03:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/933546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/46/933546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/46/933546/4"}}},"commit":{"parents":[{"commit":"7929bd422d02c45bfa9e334a3f52739f3d8ff9ef","subject":"Merge \"Expose configure_service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7929bd422d02c45bfa9e334a3f52739f3d8ff9ef"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 17:28:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 19:03:23.000000000","tz":540},"subject":"Support customizing oslo.db options","message":"Support customizing oslo.db options\n\nGnocchi uses oslo.db internally when mysql or PostgreSQL is used as\nthe indexer backend. Expose oslo.db options so that users can customize\nsome of these options for tunings.\n\nChange-Id: I248a7ab2287c8252478b25f020fc5e5f5e653813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/67d29a9f29c55d2be4341a1af546ee9599b3d429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/67d29a9f29c55d2be4341a1af546ee9599b3d429"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
