)]}'
{"id":"openstack%2Fpuppet-gnocchi~962742","triplet_id":"openstack%2Fpuppet-gnocchi~master~I7fc4340fe96be7b2bf3364eb47e2a0f2620549fd","project":"openstack/puppet-gnocchi","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-03 14:53:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fc4340fe96be7b2bf3364eb47e2a0f2620549fd","subject":"coordination: Expose parameters for package management","status":"MERGED","created":"2025-10-02 05:54:43.000000000","updated":"2025-10-03 14:58:35.000000000","submitted":"2025-10-03 14:53:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962742","meta_rev_id":"b2aff29f5b6a04e4c0cd658532faf467117fd790","_number":962742,"virtual_id_number":962742,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-03 14:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-10-02 11:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-10-03 13:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-10-03 13:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 06:09:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 11:15:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-10-03 13:43:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"0fcefb435663497de9eb62d5eb39262647082ee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 05:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5dfeac1f7a84b145df62c01b7820b765d61a81b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 06:09:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/821e96697b684c078364e9fa43c91b8f\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/061556540de0467483d9d21c7278fb33 : FAILURE in 2m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf7062a3239b4f45b4fcb7c35fcc797f : SUCCESS in 2m 22s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/a59ab981e8b543e5877a6349d62ce90e : FAILURE in 4m 33s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3785632e40a4563ab1b3c8c21efa63d : FAILURE in 4m 14s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a859d0dd674b4467922f41890ad5ceb6 : FAILURE in 4m 49s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e56a59fbbff24e64a8cb965038780e86 : FAILURE in 4m 30s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fc40a2b20834c4980d44f23b41fa0d0 : FAILURE in 9m 04s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3afd9a59d5f4dce9c7b4dd4903ce26d : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b375d2bb61374f9396f0e37d0fe25dca : FAILURE in 4m 11s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1a0a13eb6564a18a3b5c63d8bf98f50 : FAILURE in 9m 38s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d54f1528479e41dfbbf8102c4d3c72c2 : FAILURE in 8m 21s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/facbb6cf439e491ba62dcf10cb857e22 : FAILURE in 7m 16s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/faaab2726fe74ca9b37627d889c36971 : FAILURE in 5m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/994ee0a5e0f74eaaaf1684fac2d10151 : FAILURE in 4m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78e725c2030241fa860de793dfea9844 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f7e5a9c9659a446b0bc25cc7201e25656f617e3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 06:31:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"78718c41724a8dbaeab4ed3420ac3823e22e05e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 06:32:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"432d3bf96a17e41eb741192cbb3d5f78ef78491f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 07:40:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/129efd1dc67f4c5fa5c99355a6c15c90\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9f84339abc7490fb9718e7958ac289d : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90b0dd661bfc43259a4e9fb8ec746fd7 : SUCCESS in 2m 23s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/a04f3c4fccff42f59e6bf96a3a8093c7 : FAILURE in 5m 31s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05570ecdca5341b7b039ab235a005de8 : FAILURE in 4m 07s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afea311a501d476a8048d8c8febf773f : FAILURE in 4m 52s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62bfcdbbb5834e7ab79a25547da8d02f : SUCCESS in 54m 29s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9551bc846584d9fa66cfa39d307ef1a : SUCCESS in 48m 30s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8aac733cc6341eaba96b5685e9e201c : SUCCESS in 48m 42s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4c821bdcec14bcf91ce34c3afaf3b9f : SUCCESS in 1h 02m 34s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/649e30c8059343e1a0b5fa08b487c63a : SUCCESS in 46m 16s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1dd5cf893874376bf1af12c4c27a796 : SUCCESS in 38m 48s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ce886205f274147a172e78a8c310585 : SUCCESS in 12m 13s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22f67f50c45641b5aea52d089a8ef73d : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ca88223ba2645ee88df0c08c330b313 : SUCCESS in 9m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25a759e72f274c4694ddfe15c86f3eaa : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"c26d55240198b041346a9cb5c77af8ada02f6866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 08:37:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b7f671562ac7fcdc1ab0af3fbbd2d39ec57782b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 10:04:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3693ee6cd764f50991d702f50bec276\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa169f07136547a79e4454da3cca1814 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d284fb7b6ee744a28e9aad5a750f6654 : SUCCESS in 3m 05s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/bb458934dc534aa4a003e8ae967a4ae1 : SUCCESS in 4m 03s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66599ef1f5e0443e9fee2a34df473441 : SUCCESS in 5m 28s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad2d11fa6b5b4185ad0b0dfb8adf5ca0 : SUCCESS in 5m 37s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/474c087cb01745309a7460f4f1b1fb28 : SUCCESS in 53m 54s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36a11eabf49f48269a892028592fb096 : SUCCESS in 57m 00s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0b5bfd266cf4683a9ec9f60a49215b3 : SUCCESS in 1h 09m 37s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d93f6aa2dc914b4f947f57c19ef1868a : SUCCESS in 36m 36s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b60c9366f015482d9e0101a6facf8223 : SUCCESS in 1h 20m 11s\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51466ad5671848c095dcbe248719f8e0 : SUCCESS in 1h 22m 02s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d7bca5c712b468391e6b119e00034ca : SUCCESS in 8m 15s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7beb12ea01cc4256a555861686a8b646 : SUCCESS in 11m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a0ef6104eec4071b5510a74bde1b4d0 : SUCCESS in 12m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b987fc122d4f43858ae5dfbb81bd5164 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":4},{"id":"568b292f369e33cfe2bdf2f93c38af918f07d44d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-02 11:15:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a370d28ad30ef8dc4e3d9aaaf44bfb800a42aaf5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-10-03 13:43:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b790e65f374503e08d2c77aa811e6faf0721fd9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 13:44:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"750134af50d647f7d2a9ea6b4c7d6c8f91a867cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:53:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/610fbf4ec4824acc9ce133eb35672200\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2f961a8e257408c9468c94dc2d26796 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b30c3bc191db4e4da93c5c4d4fcc69f7 : SUCCESS in 2m 31s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/0af9018c856a44ce861a553de8ee0d77 : SUCCESS in 3m 54s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b54545c2f7d3430b8eca3da653a03221 : SUCCESS in 2m 58s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d831d7e73abb4c2ead8145e41befea59 : SUCCESS in 1h 05m 25s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b661fc63b4d64406818d1010f5aaf279 : SUCCESS in 40m 50s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c2e5617f4ec47d89209248bb6506fdd : SUCCESS in 16m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/275683204c6c49b4b8b8630b331adc90 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":4},{"id":"eb53ba5674b35165a86a65162ca9454938610764","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:53:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b2aff29f5b6a04e4c0cd658532faf467117fd790","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:58:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a5bb4196e1a48899d1aff61b8bf5a02\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/735021832159405580e2a093434ece18 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ad1bfa0242b01841e1f34301edf8418eb2ea5fe6","revisions":{"215a1ee1122157b146b9726b4c5d858b279374ff":{"kind":"REWORK","_number":1,"created":"2025-10-02 05:54:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/962742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/42/962742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/1"}}},"commit":{"parents":[{"commit":"003181251c08451c55ad9b72ead87b723ccfa09c","subject":"Import base parameter types for ::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/003181251c08451c55ad9b72ead87b723ccfa09c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 05:54:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 05:54:33.000000000","tz":540},"subject":"coordination: Expose parameters for package management","message":"coordination: Expose parameters for package management\n\nChange-Id: I7fc4340fe96be7b2bf3364eb47e2a0f2620549fd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/215a1ee1122157b146b9726b4c5d858b279374ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/215a1ee1122157b146b9726b4c5d858b279374ff"}]},"branch":"refs/heads/master"},"3310a8a8702a7262f904ce6505f457a8b52ea555":{"kind":"REWORK","_number":2,"created":"2025-10-02 06:31:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/962742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/42/962742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/2"}}},"commit":{"parents":[{"commit":"003181251c08451c55ad9b72ead87b723ccfa09c","subject":"Import base parameter types for ::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/003181251c08451c55ad9b72ead87b723ccfa09c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 05:54:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:31:46.000000000","tz":540},"subject":"coordination: Expose parameters for package management","message":"coordination: Expose parameters for package management\n\nChange-Id: I7fc4340fe96be7b2bf3364eb47e2a0f2620549fd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/3310a8a8702a7262f904ce6505f457a8b52ea555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/3310a8a8702a7262f904ce6505f457a8b52ea555"}]},"branch":"refs/heads/master"},"7c98f9fb266ccd826d00dfad8e3f45ab109d1124":{"kind":"REWORK","_number":3,"created":"2025-10-02 06:32:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/962742/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/42/962742/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/3"}}},"commit":{"parents":[{"commit":"003181251c08451c55ad9b72ead87b723ccfa09c","subject":"Import base parameter types for ::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/003181251c08451c55ad9b72ead87b723ccfa09c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 05:54:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:32:50.000000000","tz":540},"subject":"coordination: Expose parameters for package management","message":"coordination: Expose parameters for package management\n\nChange-Id: I7fc4340fe96be7b2bf3364eb47e2a0f2620549fd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7c98f9fb266ccd826d00dfad8e3f45ab109d1124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7c98f9fb266ccd826d00dfad8e3f45ab109d1124"}]},"branch":"refs/heads/master"},"ad1bfa0242b01841e1f34301edf8418eb2ea5fe6":{"kind":"REWORK","_number":4,"created":"2025-10-02 08:37:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/962742/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/42/962742/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/42/962742/4"}}},"commit":{"parents":[{"commit":"003181251c08451c55ad9b72ead87b723ccfa09c","subject":"Import base parameter types for ::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/003181251c08451c55ad9b72ead87b723ccfa09c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 05:54:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 08:37:18.000000000","tz":540},"subject":"coordination: Expose parameters for package management","message":"coordination: Expose parameters for package management\n\nChange-Id: I7fc4340fe96be7b2bf3364eb47e2a0f2620549fd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/ad1bfa0242b01841e1f34301edf8418eb2ea5fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/ad1bfa0242b01841e1f34301edf8418eb2ea5fe6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
