)]}'
{"id":"openstack%2Fpuppet-heat~1004744","triplet_id":"openstack%2Fpuppet-heat~master~I655768fe3efd924fe72ccce746cdcf4345e048c8","project":"openstack/puppet-heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 16:11:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I655768fe3efd924fe72ccce746cdcf4345e048c8","subject":"Deprecate support for [database] mysql_enable_nbd","status":"MERGED","created":"2026-09-09 02:44:59.000000000","updated":"2026-09-10 16:13:09.000000000","submitted":"2026-09-10 16:11:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004744","meta_rev_id":"d9233dbfc6566858f3666a3fc6b867779f07c563","_number":1004744,"virtual_id_number":1004744,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 16:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2026-09-10 10:54:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-09-10 14:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2026-09-10 14:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 04:40:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 10:54:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-09-10 14:49:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"fbc5a2ab3ad86304c8e985f1761cd39cdb2c626a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:44:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccf0a65d20c07fd09328f79b394721e7b0e10a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 04:40:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa96dbfdae10409c90c9e508f43cb8f5\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0acaa8b404ce429aa00207bbbf6db587 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b246297e1ec499d846a353bbf712d41 : SUCCESS in 3m 33s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/a5b32c22c6be4ff1ae7c7222b09d0396 : FAILURE in 9m 22s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c43f4bce8dc407ebd957ce69a254793 : FAILURE in 5m 14s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44067ce8c7d541768551f0ffd8246a70 : FAILURE in 7m 20s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fca85f212bc4a69be174242b84c2a09 : SUCCESS in 1h 00m 16s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce894e8e35d54604818e25075852cbc9 : SUCCESS in 1h 10m 18s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/12a47a3b85ca4e988ee2ddf6649392cd : SUCCESS in 59m 07s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a98522589a34ee188d6ffc1f73a8938 : FAILURE in 1h 20m 58s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/327660a0fee2470f86af366e37508622 : SUCCESS in 58m 49s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f110e58b023439ab5549bac98ae3a76 : SUCCESS in 1h 02m 05s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/38d963aa7d594e6b8e3bccc267e271e2 : FAILURE in 9m 08s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e004c69d7b284f36a084b23eb5acbcaf : SUCCESS in 15m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8c2c82fa3be4fe193d206183fe869e6 : SUCCESS in 12m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd00b1c3a084e26981b28adf00354f9 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"add246c712ca03eb0e7c9c749b694274531730bb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 16:21:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00f27fc128d9f77dec95a5b0e04c058b05b02c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:16:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c6c5098c92c4844b7e4f715cca1bfe1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0b8d8220fe3455985570d9a7182921c : SUCCESS in 4m 10s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91951f033afb446dbf3f0928cb61882f : SUCCESS in 4m 11s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/95cfe9039e404c6cb80b47e4842e71ad : SUCCESS in 9m 01s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4a52439ee4d4fbdacd2d9ed51f4f62e : SUCCESS in 8m 25s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d38196b78774c769ebd08e9e0c91c18 : SUCCESS in 8m 57s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf6a2ca61df24fefa8332247e942a7a2 : SUCCESS in 1h 07m 42s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79663d66af8944989f78ff753d151e5b : SUCCESS in 1h 08m 19s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b77e1d238c04c1f92865bfa5839a404 : SUCCESS in 1h 03m 38s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcbeabbd4a71488daa98297eee452b99 : FAILURE in 1h 21m 29s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/986ed71a7e96455b8a5be624ddeeccea : SUCCESS in 1h 06m 27s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/828861a93cae4eff803b604a496bc095 : SUCCESS in 47m 21s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ca3ea720da24315be851e2d716624e1 : SUCCESS in 21m 45s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5aacb1857a304536a4aa155853d0aaf7 : SUCCESS in 9m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d53f2116cbd4ec398749e2b66bcd09b : SUCCESS in 15m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e198c77f9ab4e8584be21566fa624cc : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a1460004f8725a0e719c52cfa32f35878ded40ba","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-10 10:54:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"156b07fd8e31222c431784e0d97087791de5db0a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-09-10 14:49:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"227f5a0f0fabf3a177c793665e197524fbbb7d8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:54:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"974ae71d7ca9addcd522a7e51ec2618f48131432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:11:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06186fd6f5b9402fa3a711b454aa95a5\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88634cdbf7894cae9257c444a839d819 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8490573e34f42249b4205591ddb3c9d : SUCCESS in 2m 42s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/49e768029ff346c384836c81fa8450af : SUCCESS in 8m 49s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7f11d4ebe794f6e9d6331d6682964a7 : SUCCESS in 8m 32s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e52647507de434090c1f16b37e4f4bf : SUCCESS in 1h 05m 54s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7488684060ac4f8dbbca1822085ded59 : SUCCESS in 1h 04m 48s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb9870c28378444e89b4d0b56972abd7 : SUCCESS in 22m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0da159ca8ced4d7a8bc3c9429daafd21 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"54e51600e3a636162d4a0743fc3396fba491a88b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:11:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9233dbfc6566858f3666a3fc6b867779f07c563","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:13:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95022849f2a4d74a2121bf235a97a99\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/673279ec0a1946ab8295f0382462f254 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"777a774f8fb47186d0bf9f1a803cf42f2ce3130c","revisions":{"777a774f8fb47186d0bf9f1a803cf42f2ce3130c":{"kind":"REWORK","_number":1,"created":"2026-09-09 02:44:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/1004744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/44/1004744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/44/1004744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/44/1004744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/44/1004744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/44/1004744/1"}}},"commit":{"parents":[{"commit":"0d0d90e1d80a9b2ec5bda1b69fb4afd84a1ca611","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0d0d90e1d80a9b2ec5bda1b69fb4afd84a1ca611"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:44:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:44:50.000000000","tz":540},"subject":"Deprecate support for [database] mysql_enable_nbd","message":"Deprecate support for [database] mysql_enable_nbd\n\nDepends-on: https://review.opendev.org/1004735\nChange-Id: I655768fe3efd924fe72ccce746cdcf4345e048c8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/777a774f8fb47186d0bf9f1a803cf42f2ce3130c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/777a774f8fb47186d0bf9f1a803cf42f2ce3130c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
