)]}'
{"id":"openstack%2Fpuppet-heat~659417","triplet_id":"openstack%2Fpuppet-heat~master~I94d00a6ef9c8949b1de8f12328a8f40dc2ebf7ca","project":"openstack/puppet-heat","branch":"master","topic":"add_openstackclient","hashtags":[],"change_id":"I94d00a6ef9c8949b1de8f12328a8f40dc2ebf7ca","subject":"Add openstackclient installation to the client class","status":"MERGED","created":"2019-05-16 01:53:23.000000000","updated":"2019-05-17 21:06:24.000000000","submitted":"2019-05-17 21:06:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659417-1558127184458-38c90c1d","meta_rev_id":"3adcdd89de8e12acd31ca597423a093e45400c14","_number":659417,"virtual_id_number":659417,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-05-17 21:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-17 10:41:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-05-17 19:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-17 19:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-05-17 19:27:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 10:41:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-17 19:22:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-17 19:27:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-05-17 21:06:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bd06007ee3974365f667c3f5d0c489fa8061c46","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-16 01:53:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd5b7378c84f15ed9227960552c8ff3052588f45","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 02:39:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/659417/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/b13a43d/ : SUCCESS in 36m 36s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/659417/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/6c1751e/ : SUCCESS in 44m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/659417/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/67a3a8f/ : SUCCESS in 43m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/659417/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/71ba332/ : SUCCESS in 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b217e0fb886d27e95971531c99b21271dac06d39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 03:43:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-lint-ubuntu-bionic/fdc0c67/ : SUCCESS in 7m 08s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-syntax-4-ubuntu-bionic/3055896/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-syntax-5-ubuntu-bionic/17dc664/ : SUCCESS in 7m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-syntax-6-ubuntu-bionic/bf6edd8/ : SUCCESS in 4m 53s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/17/659417/1/check/puppet-openstack-unit-4.8-centos-7/de0e0a9/ : SUCCESS in 11m 48s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/17/659417/1/check/puppet-openstack-unit-5.5-centos-7/c33e619/ : SUCCESS in 14m 12s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/ab58a5f/ : SUCCESS in 11m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-unit-latest-ubuntu-bionic/5b649d3/ : SUCCESS in 9m 56s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/ccd7d6b/ : FAILURE in 31m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/17/659417/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/81c9b1f/ : SUCCESS in 1h 02m 21s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/17/659417/1/check/puppet-openstack-beaker-centos-7/5509e13/ : SUCCESS in 18m 09s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/17/659417/1/check/puppet-openstack-beaker-ubuntu-bionic/f42e38c/ : FAILURE in 9m 24s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/17/659417/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/e0dbb6f/ : SUCCESS in 1h 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4424dcfc2fa49040939654481b527fcff182e208","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 01:01:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50c8d33cc7d06175dca9a6408546e327578f88fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:22:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-lint-ubuntu-bionic/618ff62/ : SUCCESS in 4m 27s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-syntax-4-ubuntu-bionic/f9e64c3/ : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-syntax-5-ubuntu-bionic/3af012c/ : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-syntax-6-ubuntu-bionic/28b3859/ : SUCCESS in 5m 02s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/17/659417/2/check/puppet-openstack-unit-4.8-centos-7/239cb39/ : SUCCESS in 18m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/17/659417/2/check/puppet-openstack-unit-5.5-centos-7/13644d3/ : SUCCESS in 12m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/47dac06/ : SUCCESS in 7m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-unit-latest-ubuntu-bionic/f6824dd/ : SUCCESS in 8m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/cca50e6/ : FAILURE in 24m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/17/659417/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/1867831/ : SUCCESS in 54m 12s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/17/659417/2/check/puppet-openstack-beaker-centos-7/706048f/ : SUCCESS in 14m 57s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/17/659417/2/check/puppet-openstack-beaker-ubuntu-bionic/d406a6f/ : FAILURE in 7m 50s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/17/659417/2/check/tripleo-puppet-ci-centos-7-undercloud-containers/b6d0cfc/ : SUCCESS in 1h 19m 12s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/17/659417/2/check/build-openstack-releasenotes/d4170f4/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1bd03d3e65dfe47614be40fecf7ed29d2aa26007","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 02:40:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/659417/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/9b7d53c/ : SUCCESS in 1h 30m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/659417/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/77562ea/ : SUCCESS in 1h 34m 58s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/659417/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/af98a8a/ : SUCCESS in 1h 36m 14s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/659417/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/7a62da5/ : SUCCESS in 1h 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b55bd9891d447806aacb43cbfa486be52dab68e9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 08:32:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e431d049e4c3b7951540480dd4cf08e5654069a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 10:03:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-lint-ubuntu-bionic/16a168d/ : SUCCESS in 6m 29s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-syntax-4-ubuntu-bionic/8173cfa/ : SUCCESS in 3m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-syntax-5-ubuntu-bionic/3ffe174/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-syntax-6-ubuntu-bionic/bbc840b/ : SUCCESS in 3m 55s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/17/659417/3/check/puppet-openstack-unit-4.8-centos-7/b594f71/ : SUCCESS in 19m 08s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/17/659417/3/check/puppet-openstack-unit-5.5-centos-7/b73500d/ : SUCCESS in 19m 40s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/7774767/ : SUCCESS in 11m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-unit-latest-ubuntu-bionic/629d2f1/ : SUCCESS in 11m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/03be3c6/ : FAILURE in 26m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/17/659417/3/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/f0d1cc0/ : SUCCESS in 58m 34s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/17/659417/3/check/puppet-openstack-beaker-centos-7/72eb9fe/ : SUCCESS in 15m 33s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/17/659417/3/check/puppet-openstack-beaker-ubuntu-bionic/1e08156/ : FAILURE in 6m 51s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/17/659417/3/check/tripleo-puppet-ci-centos-7-undercloud-containers/feb4a16/ : SUCCESS in 1h 24m 49s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/17/659417/3/check/build-openstack-releasenotes/8fda066/html/ : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c8c87302b69f925c63b7f352e3a0a94a6ad0fc9c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 10:41:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/17/659417/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/b08dea8/ : SUCCESS in 1h 11m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/17/659417/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/f22ee5e/ : FAILURE in 1h 34m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/17/659417/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/b5ef02e/ : SUCCESS in 1h 14m 54s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/17/659417/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/468af8f/ : SUCCESS in 53m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"399272fd7b54c1a06ac0d9d81978f802f237017b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-17 19:22:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05ebffc01a78c6fcdbcf80f58c7dfa3ec9a6d5b8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-17 19:23:09.000000000","message":"Patch Set 3: -Workflow\n\noops wrong button","accounts_in_message":[],"_revision_number":3},{"id":"9483247ddd70d47841f5d6f5f9894745f66d0c5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:23:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aa3dc1cf6372bf80aedc1b16f2db4dd720d964a8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-17 19:27:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"35456902928a7f3faffc4586f55f2edde9599800","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:06:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"11011a7648228d1475dc190be4f91c925f2bb80b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:06:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/17/659417/3/gate/puppet-openstack-lint-ubuntu-bionic/fda42e0/ : SUCCESS in 3m 50s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/17/659417/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/e572cca/ : SUCCESS in 6m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/17/659417/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/40fd2f5/ : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/17/659417/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/81ab3b7/ : SUCCESS in 6m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/17/659417/3/gate/puppet-openstack-unit-4.8-centos-7/3d6e6a8/ : SUCCESS in 19m 52s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/17/659417/3/gate/puppet-openstack-unit-5.5-centos-7/f8973d3/ : SUCCESS in 18m 14s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/17/659417/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/223a898/ : SUCCESS in 11m 18s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/17/659417/3/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/a7fda8e/ : SUCCESS in 1h 01m 42s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/17/659417/3/gate/puppet-openstack-beaker-centos-7/0138718/ : SUCCESS in 38m 50s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/17/659417/3/gate/puppet-openstack-beaker-ubuntu-bionic/3259588/ : FAILURE in 10m 32s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/17/659417/3/gate/tripleo-puppet-ci-centos-7-undercloud-containers/93f68df/ : SUCCESS in 1h 36m 52s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/17/659417/3/gate/build-openstack-releasenotes/72f6893/html/ : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e27652a02b3d55cc5b45a9f607dd857a1ff1a046","revisions":{"f96e45abf1708120d744c61c425bb3a2abb59435":{"kind":"REWORK","_number":1,"created":"2019-05-16 01:53:23.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/17/659417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/17/659417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/1"}}},"commit":{"parents":[{"commit":"0e380a80b1d4379ea599e43cfca3286e5c6d508f","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0e380a80b1d4379ea599e43cfca3286e5c6d508f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"subject":"Add openstackclient","message":"Add openstackclient\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I94d00a6ef9c8949b1de8f12328a8f40dc2ebf7ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/f96e45abf1708120d744c61c425bb3a2abb59435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/f96e45abf1708120d744c61c425bb3a2abb59435"}]},"branch":"refs/heads/master"},"531a2d9d25286d303cede70e49ab1e10947fe755":{"kind":"REWORK","_number":2,"created":"2019-05-17 01:01:44.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/17/659417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/17/659417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/2"}}},"commit":{"parents":[{"commit":"0e380a80b1d4379ea599e43cfca3286e5c6d508f","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0e380a80b1d4379ea599e43cfca3286e5c6d508f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 01:00:15.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I94d00a6ef9c8949b1de8f12328a8f40dc2ebf7ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/531a2d9d25286d303cede70e49ab1e10947fe755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/531a2d9d25286d303cede70e49ab1e10947fe755"}]},"branch":"refs/heads/master"},"e27652a02b3d55cc5b45a9f607dd857a1ff1a046":{"kind":"REWORK","_number":3,"created":"2019-05-17 08:32:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/17/659417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/17/659417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/17/659417/3"}}},"commit":{"parents":[{"commit":"0e380a80b1d4379ea599e43cfca3286e5c6d508f","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0e380a80b1d4379ea599e43cfca3286e5c6d508f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 08:31:17.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I94d00a6ef9c8949b1de8f12328a8f40dc2ebf7ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e27652a02b3d55cc5b45a9f607dd857a1ff1a046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e27652a02b3d55cc5b45a9f607dd857a1ff1a046"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
