)]}'
{"id":"openstack%2Fpuppet-heat~659999","triplet_id":"openstack%2Fpuppet-heat~master~Id94c1a3b6980cfedee3950e6151a0d7558457dd5","project":"openstack/puppet-heat","branch":"master","topic":"fix-client-test","hashtags":[],"change_id":"Id94c1a3b6980cfedee3950e6151a0d7558457dd5","subject":"Fix client spec test","status":"MERGED","created":"2019-05-20 03:03:53.000000000","updated":"2019-05-20 07:14:28.000000000","submitted":"2019-05-20 07:14:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659999-1558336468455-5d02b967","meta_rev_id":"30d02de770ee4d89ad820bbadb43204024d26582","_number":659999,"virtual_id_number":659999,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-20 07:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-20 05:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 05:34:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 07:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"097aa626c45900b985f846acdb5e2f7b061e40e6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 03:03:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11ce8a2e0294d983aaaa21f7af384ca5df8ed31e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:45:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/99/659999/1/check/puppet-openstack-lint-ubuntu-bionic/b4f10f4/ : SUCCESS in 4m 38s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/99/659999/1/check/puppet-openstack-syntax-4-ubuntu-bionic/461d13c/ : SUCCESS in 5m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/99/659999/1/check/puppet-openstack-syntax-5-ubuntu-bionic/1598351/ : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/99/659999/1/check/puppet-openstack-syntax-6-ubuntu-bionic/35250e0/ : SUCCESS in 3m 31s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/99/659999/1/check/puppet-openstack-unit-4.8-centos-7/2c192e1/ : SUCCESS in 13m 32s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/99/659999/1/check/puppet-openstack-unit-5.5-centos-7/0656adc/ : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/99/659999/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/f0194c9/ : SUCCESS in 11m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/99/659999/1/check/puppet-openstack-unit-latest-ubuntu-bionic/ece9212/ : SUCCESS in 9m 24s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/99/659999/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/fe97776/ : SUCCESS in 1h 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc4b66874b2df67f808ec5da0319eed5af535127","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 05:34:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e0f669990039e06362c23b83c81df493b825870c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 05:34:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22969ae68d3b2a61ee1efd9a822a1648c63a9f8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:14:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a723c647a739be9c0aae0dfa59485f1f3db8ef44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:14:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/99/659999/1/gate/puppet-openstack-lint-ubuntu-bionic/a048a0d/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/99/659999/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/f15cff3/ : SUCCESS in 3m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/99/659999/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/750baa1/ : SUCCESS in 3m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/99/659999/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/57b98d7/ : SUCCESS in 4m 25s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/99/659999/1/gate/puppet-openstack-unit-4.8-centos-7/3ac450b/ : SUCCESS in 20m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/99/659999/1/gate/puppet-openstack-unit-5.5-centos-7/c802c30/ : SUCCESS in 12m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/99/659999/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/76d8695/ : SUCCESS in 9m 24s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/99/659999/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/306029e/ : SUCCESS in 1h 28m 07s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0fdde53b1bfbd4c1bdee5cce43cf1ea669187d5","revisions":{"b0fdde53b1bfbd4c1bdee5cce43cf1ea669187d5":{"kind":"REWORK","_number":1,"created":"2019-05-20 03:03:53.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/99/659999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/99/659999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/99/659999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/99/659999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/99/659999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/99/659999/1"}}},"commit":{"parents":[{"commit":"e27652a02b3d55cc5b45a9f607dd857a1ff1a046","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e27652a02b3d55cc5b45a9f607dd857a1ff1a046"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:19.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:19.000000000","tz":480},"subject":"Fix client spec test","message":"Fix client spec test\n\nChange-Id: Id94c1a3b6980cfedee3950e6151a0d7558457dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b0fdde53b1bfbd4c1bdee5cce43cf1ea669187d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b0fdde53b1bfbd4c1bdee5cce43cf1ea669187d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
