)]}'
{"id":"openstack%2Fpuppet-heat~668804","triplet_id":"openstack%2Fpuppet-heat~master~I00f5e43146a0b937d2dfafa080f37c3601eed809","project":"openstack/puppet-heat","branch":"master","topic":"bug/1835162","hashtags":[],"change_id":"I00f5e43146a0b937d2dfafa080f37c3601eed809","subject":"Change memcache_socket_timeout to a float","status":"MERGED","created":"2019-07-03 07:19:46.000000000","updated":"2019-07-03 17:49:33.000000000","submitted":"2019-07-03 17:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668804-1562176173295-5004bf0a","meta_rev_id":"738852cbeb41b27648fff98131bb1aee0a1a6e0a","_number":668804,"virtual_id_number":668804,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-07-03 17:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-03 08:06:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-03 15:48:07.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-03 12:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-03 15:48:07.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 08:06:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 12:55:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-07-03 15:48:07.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-03 17:49:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c957cfc879a4919c95ed00739af922271fe3a0ec","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-03 07:19:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39170b64533a1600dc6bbb5a8f82c4c5c8db1153","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-03 08:06:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/04/668804/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/1c29af1/ : SUCCESS in 44m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/04/668804/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/2bb3f8a/ : SUCCESS in 40m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/04/668804/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/12fe446/ : SUCCESS in 38m 52s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/04/668804/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/213632d/ : SUCCESS in 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ea5217fb546eed0795dc000ad36b087ddbdc005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:19:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/04/668804/1/check/puppet-openstack-lint-ubuntu-bionic/a1e4084/ : SUCCESS in 5m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/04/668804/1/check/puppet-openstack-syntax-5-ubuntu-bionic/2329b82/ : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/04/668804/1/check/puppet-openstack-syntax-6-ubuntu-bionic/34f4386/ : SUCCESS in 4m 34s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/04/668804/1/check/puppet-openstack-unit-5.5-centos-7/a17789c/ : SUCCESS in 12m 08s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/04/668804/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/96e1957/ : SUCCESS in 11m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/04/668804/1/check/puppet-openstack-unit-latest-ubuntu-bionic/4693be3/ : SUCCESS in 8m 53s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/04/668804/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/0a84fe9/ : FAILURE in 30m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/04/668804/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/1f5d8b2/ : SUCCESS in 1h 02m 34s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/04/668804/1/check/puppet-openstack-beaker-centos-7/1ec5eb1/ : SUCCESS in 17m 34s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/04/668804/1/check/puppet-openstack-beaker-ubuntu-bionic/dc43b71/ : FAILURE in 8m 54s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/04/668804/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/a5ed71f/ : SUCCESS in 1h 32m 35s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/04/668804/1/check/build-openstack-releasenotes/386dcce/html/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d5829a1e95ee3e668f8b87d9375d790fa92f2a16","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-07-03 12:55:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d43624933b438847f4b614be9836b46efd7ddaf","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-03 15:48:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97ddf18093d4424fe7c86d172b117f253245226f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:48:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33fe68c9f3dd523c5876451414fb0c320349a2d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:49:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/04/668804/1/gate/puppet-openstack-lint-ubuntu-bionic/1df06cb/ : SUCCESS in 6m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/04/668804/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/d289fd8/ : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/04/668804/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/b483514/ : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/04/668804/1/gate/puppet-openstack-unit-5.5-centos-7/eb1038f/ : SUCCESS in 13m 55s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/04/668804/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/58910da/ : SUCCESS in 12m 17s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/04/668804/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/0b58a1a/ : SUCCESS in 1h 04m 42s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/04/668804/1/gate/puppet-openstack-beaker-centos-7/6a719c7/ : SUCCESS in 17m 01s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/04/668804/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/4770819/ : SUCCESS in 1h 53m 37s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/04/668804/1/gate/build-openstack-releasenotes/98f40d2/html/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"69d6bd9e9e27236f7a43a81a8a4e42c45fceb72e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:49:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"169e0ad3ce99f21f6f468cabe7d4a61460faa987","revisions":{"169e0ad3ce99f21f6f468cabe7d4a61460faa987":{"kind":"REWORK","_number":1,"created":"2019-07-03 07:19:46.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/04/668804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/04/668804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/668804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/668804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/668804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/04/668804/1"}}},"commit":{"parents":[{"commit":"82e5a7bc8b4e95f029a09722d1eecc98bd7637da","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/82e5a7bc8b4e95f029a09722d1eecc98bd7637da"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-03 07:19:27.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-03 07:19:27.000000000","tz":480},"subject":"Change memcache_socket_timeout to a float","message":"Change memcache_socket_timeout to a float\n\nsee https://review.opendev.org/#/c/519353/\n\nChange-Id: I00f5e43146a0b937d2dfafa080f37c3601eed809\nCloses-Bug: #1835162\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/169e0ad3ce99f21f6f468cabe7d4a61460faa987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/169e0ad3ce99f21f6f468cabe7d4a61460faa987"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
