)]}'
{"id":"openstack%2Fpuppet-heat~683261","triplet_id":"openstack%2Fpuppet-heat~master~I507fd2beecface5f8de35d18bfb546a653c6ef51","project":"openstack/puppet-heat","branch":"master","topic":"remove_deprecated_heat_clients_keystone_uri","hashtags":[],"change_id":"I507fd2beecface5f8de35d18bfb546a653c6ef51","subject":"Remove heat_clients_keystone_uri option","status":"MERGED","created":"2019-09-20 01:35:16.000000000","updated":"2019-09-20 21:41:38.000000000","submitted":"2019-09-20 21:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683261-1569015617334-6712634d","meta_rev_id":"954ef7a9cafd087476158dbba476c87677dcd0c8","_number":683261,"virtual_id_number":683261,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-09-20 21:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-20 02:28:42.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-09-20 07:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-09-20 18:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-09-20 18:48:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 02:28:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-20 07:40:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-09-20 18:48:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-20 21:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e3887b816f6aa085a8b2427055ce1d2b20611f1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-20 01:35:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce6b6ebe9736fe32cc97125248f61e7b1b2422be","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 02:28:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/61/683261/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/fedb5dc/ : SUCCESS in 49m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/61/683261/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e1c58d4/ : SUCCESS in 41m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/61/683261/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/c069ae6/ : SUCCESS in 52m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/61/683261/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/4e01602/ : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ea6b107885d6b53c3fee1e9db8521f0285ff797","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 04:04:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7cb083ccc9b4dfa9ff1c0e0c9d2dfbc : SUCCESS in 3m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c78f630be7c4d42af8c9924fe62c398 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25b3bdbeea9143bd8772d2022ba2e027 : SUCCESS in 3m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f22e895726844b1db820ed2428136828 : SUCCESS in 13m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0607874a95b94f8fb213cb10a49e69cf : SUCCESS in 9m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b608e6ed73f244bd825a5d9154f7e3fe : SUCCESS in 10m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/692601c7ea9447f090238c6dc78fbc73 : FAILURE in 24m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ea63e14f1c604953a433ab32e40b2abf : SUCCESS in 1h 05m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fac040ad4efa4249b084d5d68762fdea : SUCCESS in 18m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b767e6757699406eb0a94a829492f39f : FAILURE in 7m 34s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/624fe137634143aab8ab28312a727919 : SUCCESS in 2h 23m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7612230f4044e5d9b6d3cf733d31b67 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9de7de7956946e4b64d937ddc4b71457e50a1e26","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-09-20 07:40:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d63d7f4afc5b1552d5203981cb24c9ed9ee3145","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-20 18:48:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2aa0b94f45d1b26a544601c108c4242f48fd02d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 18:48:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54e3e6bdcf14c75f4252f62a29b724b009852691","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:40:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c28edc1aca477b583f4b7a492ca4d52ddf85198","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:40:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4412690f16ba4e33b2703ec30efa13b6 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2afa6fed04fa495fbf1d5f5e0b0de5e4 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/083105081d5f4da8ab1d44ee10d6638d : SUCCESS in 4m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/80422ab1ee234592bd0ba12de1d87a9a : SUCCESS in 26m 22s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a996a504c073416585b9fc14086513f7 : SUCCESS in 13m 18s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a9ead4bd5872415b85b86da27b19dde5 : SUCCESS in 1h 07m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b8908c537e21470c8a5f89ce76cb3295 : SUCCESS in 20m 37s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ea5585baa38b4a5b8964067098c4aa94 : SUCCESS in 2h 35m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/962febe0c49648e6b34a971c96f196ef : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ba3f250c34caaf141d371f831d63d2d87e71ac7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 21:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8e74398fe524d8487ea8c10b9dc8f4d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c35662238f05765f64df1ca9670d73c1ab6a006","revisions":{"4c35662238f05765f64df1ca9670d73c1ab6a006":{"kind":"REWORK","_number":1,"created":"2019-09-20 01:35:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/61/683261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/61/683261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/683261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/683261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/683261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/61/683261/1"}}},"commit":{"parents":[{"commit":"e96248bca390bd1aca7d5eeaf5d0a679b61910ec","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e96248bca390bd1aca7d5eeaf5d0a679b61910ec"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-20 01:34:27.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-20 01:34:27.000000000","tz":480},"subject":"Remove heat_clients_keystone_uri option","message":"Remove heat_clients_keystone_uri option\n\nThe heat_clients_keystone_uri parameter has been deprecated for one year.\nWe can remove it.\n\nChange-Id: I507fd2beecface5f8de35d18bfb546a653c6ef51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4c35662238f05765f64df1ca9670d73c1ab6a006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4c35662238f05765f64df1ca9670d73c1ab6a006"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
