)]}'
{"id":"openstack%2Fpuppet-heat~686648","triplet_id":"openstack%2Fpuppet-heat~stable%2Ftrain~I8e1029077ba3270e95c332e80f4b62d40a389068","project":"openstack/puppet-heat","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"I8e1029077ba3270e95c332e80f4b62d40a389068","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-10-04 13:26:39.000000000","updated":"2019-10-10 19:08:32.000000000","submitted":"2019-10-10 19:08:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686648-1570734512980-0f5a3271","meta_rev_id":"aa67a454a9f11177cf22c134d0c43ccc0d5a1022","_number":686648,"virtual_id_number":686648,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 19:08:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-10-04 14:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-10-08 18:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-10-08 18:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 14:34:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-08 18:00:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-10 19:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11f5d7e2008da306b22c7627e5bc25f34bbdddde","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-04 13:26:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d79765d0b5fd79094abcd2cc5ac38b072e5d3ad","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-04 14:34:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"678f8567994eaefa89ba504f5ff4fb9790d2df40","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-08 18:00:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4f1b711ca750fac19e663c8d587a8231e6755f7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 21:08:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2370d79d24ba4d9c858129bb6f1a0131 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6920bc46a574451dbc4d4159cc9c821e : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8612e87e88634fb29acc52dda9e319a4 : SUCCESS in 11m 35s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9aabae7449f349d3a6c7bffcc156842b : FAILURE in 21m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6d1639d0eb1b4a789f9ef8073d2d8b62 : FAILURE in 1h 21m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a3b4b559ca9549d1b393a28f74311a08 : SUCCESS in 19m 14s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfd754b2be8e4886b26c0689deb3206a : FAILURE in 8m 20s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0f2b3007450d4948a8a7430e49ed9979 : SUCCESS in 2h 55m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81fe1a0900d5582cbae36d6d2a5f43e0a400b8cc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-09 19:26:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f2f2f05a9141ec3cf4c90dc9f951d0abd2669e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 21:37:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c1449b1e91b41c5a7826da4f167c137 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/554731f232ad4fd4aed37502e8d11d6e : SUCCESS in 4m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cd2b7c99480b4bcfb5f3903af8a2c0dd : SUCCESS in 14m 11s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b018290ead3449c84f84b44657ac15b : FAILURE in 22m 48s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2642bd7486c246ca8c62eda84f7e5152 : SUCCESS in 1h 11m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/08461f8cf2a74ccdb6cbcc05ef191979 : SUCCESS in 18m 28s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3d3c17101cb413d89cb7f747715fb45 : FAILURE in 8m 25s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3d445a2d5a554011bd35dc29bd404311 : SUCCESS in 2h 07m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a296e928fba200088f9cfef42660561f0012ae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 21:37:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e45443725118f6caeecec8ed25fec607248cce22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 23:54:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57719830c7b04eda9103c503b302ecf5 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b26950454594541b8aa3f8aa05eab8b : SUCCESS in 4m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4067aaa6cf82436bbedfd6aa10fdb3db : FAILURE in 7m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5dba26a3a18f4eac9d86d15878ccbacf : SUCCESS in 1h 07m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/56f54939a6d34511ad9f1ba3a9a88ccb : FAILURE in 8m 10s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e92a41a6a12e4d67a4207d0995b653e5 : SUCCESS in 2h 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"077b8576171a2fd5f58471559456a6836ab63ab1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-10 14:34:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"143d749dd67ea158f9008b34e4f16d572857e434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 16:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c87c6f3fc24490cac83ecd2c04dc114 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f690cc481c2d49a69877f00a87e7baea : SUCCESS in 4m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7e5db56b20cd471a9ac78d5f132da437 : SUCCESS in 23m 49s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c370224c9284a95b7fc4c094b919bfa : FAILURE in 21m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/315a884857fd444bade0bdd3768e5f47 : SUCCESS in 1h 28m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9916c5a42f3f49dd9e7c66ee84e4235a : SUCCESS in 24m 13s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7e7751b418848f68f00730c8ce09ae4 : FAILURE in 9m 23s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b5cee7be3e7413984cb47a9f7345e25 : SUCCESS in 2h 08m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b720fbe7197f66e3c3b5d9bee5d32f8be6800487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 16:53:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e9922b032ca1f54f125abf9dc5f51eb20bcaa04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:08:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1dc6ac2189585a989bd8aa71378c1d71ab75b201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:08:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20a42cf99cf44e59b7ef6168f684873d : SUCCESS in 6m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f4daf90270242daa607b5f6a97d6571 : SUCCESS in 4m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/88054fe579954b6fa6ceeb13731fa446 : SUCCESS in 18m 44s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/035ba7cb6b7b4bb28bf53a84242cd8a5 : SUCCESS in 1h 07m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f8944e4b53a04c989caccbba043204a2 : SUCCESS in 17m 32s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/70c17ed195e04cac8580dc6d12487120 : SUCCESS in 2h 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce50e591facf6f7b31eed6cfc9ddd898eebb110b","revisions":{"ce50e591facf6f7b31eed6cfc9ddd898eebb110b":{"kind":"REWORK","_number":1,"created":"2019-10-04 13:26:39.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/48/686648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/48/686648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/48/686648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/48/686648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/48/686648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/48/686648/1"}}},"commit":{"parents":[{"commit":"95c845721eac1087beb50bbd5b00dc10660d5203","subject":"Merge \"Update the constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/95c845721eac1087beb50bbd5b00dc10660d5203"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-04 13:26:35.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-04 13:26:35.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: I8e1029077ba3270e95c332e80f4b62d40a389068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ce50e591facf6f7b31eed6cfc9ddd898eebb110b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ce50e591facf6f7b31eed6cfc9ddd898eebb110b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
