)]}'
{"id":"openstack%2Fpuppet-heat~698007","triplet_id":"openstack%2Fpuppet-heat~master~Ibd3999a394c8827efa7a6034e69e55858d2e44ed","project":"openstack/puppet-heat","branch":"master","hashtags":[],"change_id":"Ibd3999a394c8827efa7a6034e69e55858d2e44ed","subject":"Make heat config client_retry_limit configurable","status":"MERGED","created":"2019-12-09 12:40:03.000000000","updated":"2019-12-11 08:01:06.000000000","submitted":"2019-12-11 08:00:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698007-1576051201069-0c0e24f5","meta_rev_id":"dbe1c71d8054a221038547b106b6e98ee785452b","_number":698007,"virtual_id_number":698007,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 08:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-09 13:31:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-12-11 05:51:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-12-09 15:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-12-11 05:51:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 13:31:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-09 15:42:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-11 05:51:44.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-12-11 08:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9493318ef2c6bf948c910de38d830f1d2617e5ba","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2019-12-09 12:40:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80428eb69d09472951f68a030757e976ea604bc4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-09 13:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/07/698007/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/1fdcc0f/ : SUCCESS in 39m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/07/698007/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/26f9389/ : SUCCESS in 49m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/07/698007/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/5f49b28/ : SUCCESS in 44m 45s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/07/698007/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/6a73d1d/ : SUCCESS in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6530532d235b77ffeb8fc0f41dfe1b8dfd8aec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 14:48:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a80f171c274149a59c9f4ae99cc74765 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa9bc69808bf4735bb6cc96eeb437faf : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6f0d5876d984258af26b82f1a67ca1e : SUCCESS in 5m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2390c11e6621409f884404399370b280 : SUCCESS in 27m 20s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c39f6bdab624fdaacee75d1a206ceef : SUCCESS in 11m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4267f5377d6a4aeda1af8c0d7b3beb08 : SUCCESS in 11m 48s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81025b1b6bd74423b218767b6441026f : SUCCESS in 56m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2b4ebd6adba44fa5abcd9259fc8d306e : SUCCESS in 1h 08m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/39485355db064445a86df16cd0bfd0ba : SUCCESS in 17m 48s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9a5c73f611443038b4301f034049b44 : FAILURE in 24m 07s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b94affc3e97e4cc38c0de920a85eb687 : SUCCESS in 2h 03m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62e5decd47954f57992efa694577521e : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"00b4e2178d6096bbff4c1b7d3fcbcf5c0643f24c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-09 15:42:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d07788a52215f4cabc7605230f372a67a83f2934","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-11 05:51:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"937bb94bb98385dcc0c80ab2bbe6bbb7fc4bd9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 05:51:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62d7aba471a5e5b600ebf3930fe81101662a7f64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:00:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f2638cb7a2e41309426d7d94ee24403 : SUCCESS in 5m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78c3a9a2389742fda977b2f8c8e7a836 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e31098f38c84657bf491eefd397739d : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a9dc498e3cf946869ef74daba2d723ec : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f70d043b1a640e989103e4faf507c51 : SUCCESS in 12m 12s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f3bf3affca4648b0a981d08bf934595c : SUCCESS in 1h 03m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5458fe1544c44576ba403920040e27f7 : SUCCESS in 18m 06s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/92c63936812647c3a62001fb95280ecc : SUCCESS in 2h 04m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a618663e8f145b49e11f4e4afe54285 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"706598f200258f620bb45399eb36f0d83e3740e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:00:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dbe1c71d8054a221038547b106b6e98ee785452b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:01:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47dcb8021c6e43a69359d378c1e0c0e4 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"876a4bddfe4f6547d1a5517afb923f3f691b3f78","revisions":{"876a4bddfe4f6547d1a5517afb923f3f691b3f78":{"kind":"REWORK","_number":1,"created":"2019-12-09 12:40:03.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/07/698007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/07/698007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/07/698007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/07/698007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/07/698007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/07/698007/1"}}},"commit":{"parents":[{"commit":"b4ab38b179190575a3cee7a337a59ab941081a1f","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b4ab38b179190575a3cee7a337a59ab941081a1f"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-12-09 11:35:53.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-12-09 12:39:56.000000000","tz":330},"subject":"Make heat config client_retry_limit configurable","message":"Make heat config client_retry_limit configurable\n\nThis adds a new parameter to configure client_retry_limit config\nto a value higher than the default.\n\nChange-Id: Ibd3999a394c8827efa7a6034e69e55858d2e44ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/876a4bddfe4f6547d1a5517afb923f3f691b3f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/876a4bddfe4f6547d1a5517afb923f3f691b3f78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
