)]}'
{"id":"openstack%2Fpuppet-heat~698162","triplet_id":"openstack%2Fpuppet-heat~master~Ib70e230c964ecb720d06d8f550f6a308f6b05f7b","project":"openstack/puppet-heat","branch":"master","topic":"prep-release-ussuri-1","hashtags":[],"change_id":"Ib70e230c964ecb720d06d8f550f6a308f6b05f7b","subject":"Prepare Ussuri M1","status":"MERGED","created":"2019-12-10 06:25:50.000000000","updated":"2019-12-10 16:41:08.000000000","submitted":"2019-12-10 16:41:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698162-1575996068546-7edb6769","meta_rev_id":"e795b395a3621694406f0af53594ca849e8071b4","_number":698162,"virtual_id_number":698162,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 16:41:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-12-10 16:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-12-10 16:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 16:34:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-10 16:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d98d4fc33b12f8fbaa1820ca48a306e5296bb137","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-10 06:25:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99e2deadd3abe9367733a049d3e8948ed801eeac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 06:36:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ecf98009db3647e092f210959fc6be3c : SUCCESS in 2m 11s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2091b018eea148db8d98f80bb166ffc4 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b86988bab128033eb3c060df3a37991d87a59855","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-10 16:34:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3d631db74360387280700ba43b3d4bd2103c60a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:34:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef6da53022175b1f540506ef3d53c4045dfd725b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:41:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c5c75199e52e447fdd17a44b170556f8d2eee75a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:41:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/172778246ac2458eaf0bb683cc041589 : SUCCESS in 2m 58s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4067b0fa7e54ee9b42be3d05db80f26 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3024129a80baeb7be67d111619663fac8adec3aa","revisions":{"3024129a80baeb7be67d111619663fac8adec3aa":{"kind":"REWORK","_number":1,"created":"2019-12-10 06:25:50.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/62/698162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/62/698162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/62/698162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/62/698162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/62/698162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/62/698162/1"}}},"commit":{"parents":[{"commit":"5f6411c3dd3a82b9efadbc3cb6a363bf8a044fb7","subject":"Convert all class usage to relative names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/5f6411c3dd3a82b9efadbc3cb6a363bf8a044fb7"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-10 06:24:48.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-10 06:24:48.000000000","tz":480},"subject":"Prepare Ussuri M1","message":"Prepare Ussuri M1\n\nUpdate the version metadata for Ussuri milestone M1\n\nChange-Id: Ib70e230c964ecb720d06d8f550f6a308f6b05f7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/3024129a80baeb7be67d111619663fac8adec3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/3024129a80baeb7be67d111619663fac8adec3aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
