)]}'
{"id":"openstack%2Fpuppet-heat~698359","triplet_id":"openstack%2Fpuppet-heat~master~I4b113bbbcc1b6168dc8f19cdf52e18ced337a987","project":"openstack/puppet-heat","branch":"master","topic":"update-stdlib-version-6.1","hashtags":[],"change_id":"I4b113bbbcc1b6168dc8f19cdf52e18ced337a987","subject":"Fix stdlib version in metadata","status":"MERGED","created":"2019-12-11 03:53:46.000000000","updated":"2019-12-11 05:48:27.000000000","submitted":"2019-12-11 05:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698359-1576043307899-5614962f","meta_rev_id":"a50bca7f11cd81dbc12a837fb9409261b3ed7f26","_number":698359,"virtual_id_number":698359,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 05:48:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-12-11 05:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-12-11 05:41:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 05:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50232ba8991bdb89ec3ef917f1632246e0774e59","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-11 03:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"043a2ce11967539b3f37541c57e357bb4149b56e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 04:01:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/8a3e86f0ab4e4036839f458fcd0bdeee : SUCCESS in 2m 06s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b50e6696ace468cacc927857e039eb6 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6e8bf22de95859df30f97e302a1ae2aa527a0176","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-11 05:41:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e3a3b8a7f37e50b228f785fce2f6bdca50bbbc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 05:42:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6d2f3db70dfc4932f3171dc3ba0c9de7f1a800d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 05:48:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ccfc18ce937094b249417f2852e1593c1c1a6077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 05:48:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/3dd7564dfdc84807a9c1c5f3771fb35a : SUCCESS in 1m 43s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdbee99fff00487cbff555c655c1e13b : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"580ae13cb51b7a1a3e3570d78d9aeb47ed7205a5","revisions":{"580ae13cb51b7a1a3e3570d78d9aeb47ed7205a5":{"kind":"REWORK","_number":1,"created":"2019-12-11 03:53:46.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/59/698359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/59/698359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/59/698359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/59/698359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/59/698359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/59/698359/1"}}},"commit":{"parents":[{"commit":"3024129a80baeb7be67d111619663fac8adec3aa","subject":"Prepare Ussuri M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/3024129a80baeb7be67d111619663fac8adec3aa"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-11 03:53:03.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-11 03:53:03.000000000","tz":480},"subject":"Fix stdlib version in metadata","message":"Fix stdlib version in metadata\n\npuppetlabs-stdlib is now using the 6.1.0 version, so we should fix the\nstdlib version in metadata.\n\nChange-Id: I4b113bbbcc1b6168dc8f19cdf52e18ced337a987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/580ae13cb51b7a1a3e3570d78d9aeb47ed7205a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/580ae13cb51b7a1a3e3570d78d9aeb47ed7205a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
