)]}'
{"id":"openstack%2Fpuppet-heat~704192","triplet_id":"openstack%2Fpuppet-heat~stable%2Ftrain~I7da55899b9108daced7adb8f82bdb58fcf97aa09","project":"openstack/puppet-heat","branch":"stable/train","topic":"custom-stable/train","hashtags":[],"change_id":"I7da55899b9108daced7adb8f82bdb58fcf97aa09","subject":"Add vhost_custom_fragment option to wsgi/apache_api","status":"MERGED","created":"2020-01-24 19:58:53.000000000","updated":"2020-01-25 15:18:04.000000000","submitted":"2020-01-25 15:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704192-1579965409896-17819b5d","meta_rev_id":"0d868e356261b80b37d3d56f856bbe9dcedfb98b","_number":704192,"virtual_id_number":704192,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-25 15:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-01-24 20:00:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-01-25 13:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-01-25 13:43:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 20:00:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-01-25 13:43:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-01-25 15:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"919552014ced91563f520a19fa2d19a32daf59c3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-01-24 19:58:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e93139864a57597cd0287ce317ec90ebfabff776","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-24 20:00:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa6445f6b783a40d22d0fdae4cd1b63e0897ec92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 21:42:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0702c20e6834503b6deb03c1d914e69 : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5c832adb4c24b089f4dcb2383d049f0 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c22544cb79d1424faf1a59854fac066a : SUCCESS in 5m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fa6e145accab4ba69b058eed74921429 : SUCCESS in 18m 04s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfabe3fa767b4b039ea91c6561e43f06 : SUCCESS in 36m 45s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/80977d3b2f2b4d76b7be5e8bcbb9b7bf : SUCCESS in 1h 29m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cd0ee806a03b4beebc899119eb21ca9c : SUCCESS in 19m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6ef59bc74af4f6185937ed084764b8f : FAILURE in 20m 06s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/455a45942a134e3187b87fe6e44d4b84 : SUCCESS in 1h 38m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1acb1fdd1b14607882510e9cb1fef7f : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"92f82770f6c0d57b275ace52c1e48ee2476751c1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-01-25 13:43:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2321c543c8a3bed8b91000196abce43a8b656b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 13:43:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbae920c05c649fe709398bf5fdeab3576674761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 15:16:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6c76a22a20a4b22bd183869e8435d76 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aa9ecd6160a477288668304ea409bae : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36a78d7083cf46938ebd4a0162237ead : SUCCESS in 4m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ecf20d004e9a4bd5b524504762d2549f : SUCCESS in 14m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5f001621e2a844c6b74217355e35391a : SUCCESS in 1h 18m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/475098f3278548e891c250d702f5dd8f : SUCCESS in 19m 37s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/eb4e423b324941f38e3e3e2da8d950da : SUCCESS in 1h 31m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/154f0e9d2eac4c8493bb017509510e95 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3ba2ca49bab21a84f8da5c8fc9c48ac64a34a5a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 15:16:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84b007ade6d0a6e7267a23a1df4132e12f1574d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 15:18:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d373190b5f44c8ade876d1bdbe4740 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a18c968c5f48b885bba294bc017d5ec498337f2","revisions":{"6a18c968c5f48b885bba294bc017d5ec498337f2":{"kind":"REWORK","_number":1,"created":"2020-01-24 19:58:53.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/92/704192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/92/704192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/92/704192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/92/704192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/92/704192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/92/704192/1"}}},"commit":{"parents":[{"commit":"68b7e29cacbcec03cac5d50e78112fcffb830a04","subject":"Merge \"Switch to Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/68b7e29cacbcec03cac5d50e78112fcffb830a04"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-01-22 13:59:50.000000000","tz":-300},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-01-24 19:58:53.000000000","tz":0},"subject":"Add vhost_custom_fragment option to wsgi/apache_api","message":"Add vhost_custom_fragment option to wsgi/apache_api\n\nChange-Id: I7da55899b9108daced7adb8f82bdb58fcf97aa09\n(cherry picked from commit 9fce8d211ea4375f4a22e1040adb88575fa4c1a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/6a18c968c5f48b885bba294bc017d5ec498337f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/6a18c968c5f48b885bba294bc017d5ec498337f2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
