)]}'
{"id":"openstack%2Fpuppet-heat~709668","triplet_id":"openstack%2Fpuppet-heat~master~I41964d086f197d2a4698e89c7d490c96ccce82ce","project":"openstack/puppet-heat","branch":"master","topic":"heat_db_retries","hashtags":[],"change_id":"I41964d086f197d2a4698e89c7d490c96ccce82ce","subject":"Support retrying of disconnected DB transactions","status":"ABANDONED","created":"2020-02-25 09:04:17.000000000","updated":"2020-03-12 03:26:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0087066bf1f17aaad3aa706b87137f650015651","_number":709668,"virtual_id_number":709668,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-25 11:15:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-10 12:22:30.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":-1,"date":"2020-03-10 06:10:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2020-03-09 14:53:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"date":"2020-03-06 15:01:30.000000000","_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"date":"2020-02-25 15:36:36.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 11:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-25 15:36:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-06 15:01:30.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2020-03-09 14:53:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-10 06:10:37.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"7c6a6ac5d23dda4093d14930549a2be9f3216ac2","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-02-25 09:04:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20b02c83e580362548e25c900140f08aabf770a1","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-02-25 09:04:37.000000000","message":"Topic set to heat_db_retries","accounts_in_message":[],"_revision_number":1},{"id":"2e9b95eb6853cba19b0b76ba48144344271e224f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-02-25 09:05:43.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a44c8828e00f653c93aa7a94d278c62b6030d739","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-02-25 09:06:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb8e79194252ed4f63659dccf0849e83bfb57cbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:15:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76f46d4ad9a346babe8807b5ac9d3195 : SUCCESS in 6m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc065ca4cebc46e6921b3abe252708a2 : SUCCESS in 6m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f437c13bf80f494c954bef974f7f5a25 : SUCCESS in 5m 30s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/090b964a34fe417c94cdef49281455e1 : SUCCESS in 12m 21s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7565c3a007446d7ba23752dd389b53a : SUCCESS in 11m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ade0c84944b459587840b80ff32b418 : SUCCESS in 11m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74d801dfa2ce4e42a9dab254159b95c5 : SUCCESS in 46m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/00117420bed54a1c8a57ab05c9500df6 : SUCCESS in 1h 10m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fd564f4b1d6e402da34009214ce218f6 : SUCCESS in 1h 06m 55s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5c2b8d84a1294bf2b5d2381a2c5d2f09 : SUCCESS in 21m 03s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4ec997daa254a92b55d83fa27099bfa : FAILURE in 18m 37s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/367681be2930422c8e7bb72b36dba7ea : SUCCESS in 1h 54m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccc95a50b381475bae2bac96b81ae09f : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5dec46b927e64f9461133da568e9d1320439c377","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-02-25 15:36:14.000000000","message":"Patch Set 3:\n\nnote that currently in Heat code there is nothing that can use of that option. I\u0027ll try to improve that situation in the linked story scope.","accounts_in_message":[],"_revision_number":3},{"id":"8287c5a9cf478212e89fda301b73741befd85f6b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-03-05 14:57:56.000000000","message":"Patch Set 3:\n\nwhat\u0027s the rationale for this to be optional ?","accounts_in_message":[],"_revision_number":3},{"id":"ed742d99a48c6a180b8f723c27e793a5e5c9ca04","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-06 14:12:55.000000000","message":"Patch Set 3:\n\nI think there is no cases for deploying Heat without that capability :)\nShall I hardcode it?","accounts_in_message":[],"_revision_number":3},{"id":"4686683e349352a295263cc7c96cd82accc4e609","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-03-06 15:01:30.000000000","message":"Patch Set 3:\n\nother serivces like Nova and Neutron do retries unconditionally, making it an option is not the current convention.","accounts_in_message":[],"_revision_number":3},{"id":"e945256c62bee94cb1826164abb7522fa246cedd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 14:53:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"19a5ea05b60b8f59f511903ef1a5d7e19bbeccdc","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-03-10 06:10:37.000000000","message":"Patch Set 3: Code-Review-1\n\nAFAIK SQLAlchemy automatically reconnects when the connection is lost. \n\nLooks like this flag only has meaning[1] when the @safe_for_db_retry is used. If we\u0027re using @wrap_db_retry (which db uses atm for deadlocks), using retry_on_disconnect\u003dTrue[2] with @wrap_db_rety should be enough, unless I miss something.\n\nTherefore I\u0027m not sure if this patch is necessary. Also, I\u0027ve not completely following all changes/discussion in https://review.opendev.org/#/c/711075/ where it seems we\u0027re not using @safe_for_db_retry and also it seems zzzeek has number of concerns too. Till we sort those out there is no point merging this.\n\n\n[1] https://github.com/openstack/oslo.db/blob/bf5b45ebba24b4c8737c3197f2cd65f84fd2448c/oslo_db/api.py#L275\n[2] https://github.com/openstack/heat/blob/master/heat/db/sqlalchemy/api.py#L98","accounts_in_message":[],"_revision_number":3},{"id":"b752826cab67bbd693e84402a7075f2024049933","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-10 12:22:30.000000000","message":"Patch Set 3:\n\n\u003e Looks like this flag only has meaning[1] when the @safe_for_db_retry is used. If we\u0027re using @wrap_db_retry (which db uses atm for deadlocks), using retry_on_disconnect\u003dTrue[2] with @wrap_db_rety should be enough, unless I miss something.\n\n@Rabi, I\u0027m not an Oslo.db expert, but quick looking into its code showed to me the both ways (via @safe wrapper and/or @wrap) are the same and by default there is no disconnects on retrying, whatever of those two methods one uses","accounts_in_message":[],"_revision_number":3},{"id":"bbb32a6c2facbb8f6b204cb8fdb131b3510ba69a","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-03-10 13:02:09.000000000","message":"Patch Set 3:\n\n\u003e @Rabi, I\u0027m not an Oslo.db expert, but quick looking into its code showed to me the both ways (via @safe wrapper and/or @wrap) are the same and by default there is no disconnects on retrying, whatever of those two methods one uses\n\n@Bodgan, I probably don\u0027t understand what you mean by \u0027disconnects on retyring\u0027.\n\nWhat I\u0027m trying to say is\n\n- \u0027use_db_reconnect\u0027 flag only has significance when @safe_for_db_retry decorator is used[1]\n\n- With @wrap_db_retry decorator which us used by heat atm, if you add retry_on_disconnect\u003dTrue, it would retry on DBConnectionError[2].\n\nSo I am not sure I understand why we\u0027re adding this flag, if we\u0027re not using @safe_for_db_retry decorator.\n\n[1] https://github.com/openstack/oslo.db/blob/master/oslo_db/api.py#L272-L285\n\n[2] https://github.com/openstack/oslo.db/blob/master/oslo_db/api.py#L125-L126","accounts_in_message":[],"_revision_number":3},{"id":"d7ff6fd46fad61329bd386b8c55ce50595b9831e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-10 14:07:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"ddfb62d2e8942825c2441529a5b48927e11af8b7","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-11 15:52:02.000000000","message":"Patch Set 3:\n\n\u003e AFAIK SQLAlchemy automatically reconnects when the connection is lost. \n\n@Rabi, see oslo.db unit tests illustrating this better https://review.opendev.org/#/c/711221/3/oslo_db/tests/test_api.py :\n\nIf we don\u0027t set it to retry on disconnects for the wrapper explicitly it won\u0027t retry.","accounts_in_message":[],"_revision_number":3},{"id":"4aee4755ccc50968295d2ae784f062f7bdaa4739","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-11 15:53:21.000000000","message":"Patch Set 3:\n\nthis is abandoned because there is no use of it in Heat at the moment.\nIf we have any progress with the tasks in the story, I\u0027ll restore this.\n\nUntil then this parameter would remain a disconnected knob :)","accounts_in_message":[],"_revision_number":3},{"id":"a0087066bf1f17aaad3aa706b87137f650015651","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-03-12 03:26:20.000000000","message":"Patch Set 3:\n\n\u003e If we don\u0027t set it to retry on disconnects for the wrapper explicitly it won\u0027t retry.\n\nSure, I think I mentioned that in my previous comment, quoting again..\n\n\"With @wrap_db_retry decorator which is used by heat atm, if you add retry_on_disconnect\u003dTrue, it would retry on DBConnectionError[[2]\"\n\nI\u0027ve pushed a WIP patch https://review.opendev.org/#/c/712606/ that you can try and see if it helps with your issues.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e068e220347fde520b0b5a1da434055c3d29af33","revisions":{"7eba05a4b3bc0f3a3f33f63abd567d58fd4b9bba":{"kind":"REWORK","_number":1,"created":"2020-02-25 09:04:17.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/68/709668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/68/709668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/1"}}},"commit":{"parents":[{"commit":"6184602d863a937fb32cd07557d4e54d2391440e","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/6184602d863a937fb32cd07557d4e54d2391440e"}]}],"author":{"name":"Bogdan Dobrelya","email":"bogdando@mail.ru","date":"2020-02-25 08:59:01.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bogdando@mail.ru","date":"2020-02-25 09:04:15.000000000","tz":60},"subject":"Support retrying of disconnected DB transactions","message":"Support retrying of disconnected DB transactions\n\nOslo DB has a special decorator for API methods to indicate either\nit is a safe for re-connection to database or not. This only works\nin tandem with [database]/use_db_reconnect\u003dTrue.\n\nAdd that use_db_reconnect in order to enable such behavior by default.\nThis change has no any effect until there is other changes in Heat code\nthat mark API methods as safe for re-connection to database.\n\nStory: 2007328\ntask: 38849\n\nChange-Id: I41964d086f197d2a4698e89c7d490c96ccce82ce\nSigned-off-by: Bogdan Dobrelya \u003cbogdando@mail.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7eba05a4b3bc0f3a3f33f63abd567d58fd4b9bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7eba05a4b3bc0f3a3f33f63abd567d58fd4b9bba"}]},"branch":"refs/heads/master"},"a8544e58cd5fcebba810734fad137cf673b61056":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-25 09:05:43.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/68/709668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/68/709668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/2"}}},"commit":{"parents":[{"commit":"6184602d863a937fb32cd07557d4e54d2391440e","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/6184602d863a937fb32cd07557d4e54d2391440e"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2020-02-25 08:59:01.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2020-02-25 09:05:22.000000000","tz":60},"subject":"Support retrying of disconnected DB transactions","message":"Support retrying of disconnected DB transactions\n\nOslo DB has a special decorator for API methods to indicate either\nit is a safe for re-connection to database or not. This only works\nin tandem with [database]/use_db_reconnect\u003dTrue.\n\nAdd that use_db_reconnect in order to enable such behavior by default.\nThis change has no any effect until there is other changes in Heat code\nthat mark API methods as safe for re-connection to database.\n\nStory: 2007328\ntask: 38849\n\nChange-Id: I41964d086f197d2a4698e89c7d490c96ccce82ce\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/a8544e58cd5fcebba810734fad137cf673b61056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/a8544e58cd5fcebba810734fad137cf673b61056"}]},"branch":"refs/heads/master"},"e068e220347fde520b0b5a1da434055c3d29af33":{"kind":"REWORK","_number":3,"created":"2020-02-25 09:06:52.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/68/709668/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/68/709668/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/68/709668/3"}}},"commit":{"parents":[{"commit":"6184602d863a937fb32cd07557d4e54d2391440e","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/6184602d863a937fb32cd07557d4e54d2391440e"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2020-02-25 08:59:01.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2020-02-25 09:06:43.000000000","tz":60},"subject":"Support retrying of disconnected DB transactions","message":"Support retrying of disconnected DB transactions\n\nOslo DB has a special decorator for API methods to indicate either\nit is a safe for re-connection to database or not. This only works\nin tandem with [database]/use_db_reconnect\u003dTrue.\n\nAdd that use_db_reconnect in order to enable such behavior by default.\nThis change has no any effect until there is other changes in Heat code\nthat mark API methods as safe for re-connection to database.\n\nStory: 2007328\ntask: 38849\n\nChange-Id: I41964d086f197d2a4698e89c7d490c96ccce82ce\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e068e220347fde520b0b5a1da434055c3d29af33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e068e220347fde520b0b5a1da434055c3d29af33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
