)]}'
{"id":"openstack%2Fpuppet-heat~714880","triplet_id":"openstack%2Fpuppet-heat~master~Ibd40fed67872c502f54f12c07ebe20fa47089aa5","project":"openstack/puppet-heat","branch":"master","topic":"bug/1868511","hashtags":[],"change_id":"Ibd40fed67872c502f54f12c07ebe20fa47089aa5","subject":"Deprecate min_pool_size option","status":"MERGED","created":"2020-03-25 08:23:18.000000000","updated":"2020-03-25 20:12:25.000000000","submitted":"2020-03-25 20:10:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714880-1585167023570-b25ac13a","meta_rev_id":"7ec44b3813fc509d4eed1562ce3c28d7883224dd","_number":714880,"virtual_id_number":714880,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 20:10:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-03-25 17:29:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-25 19:02:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 17:29:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-25 19:02:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-25 20:10:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df6803f1ab969efc2a49e7991c11e4ab7ec1cc40","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-25 08:23:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e71005612cb61dec595bd3bc2df544130e1e498d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 09:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55cd04a53a244ba0b8d771fa1e91d3d3 : SUCCESS in 6m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82e31de02f3141808ebce9ed2769115c : SUCCESS in 6m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bba32a3c9074db1bd21d2c9b35ffd86 : SUCCESS in 7m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/92f88111dd58470493df296c1d0a33bd : SUCCESS in 21m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fe459880bf6c4af6a40c65f7e006bf76 : SUCCESS in 17m 33s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/5d8cc5fb82184d77abd46f7cd2a8c5f1 : SUCCESS in 18m 10s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/862b1513f1644c59a5c84fc74f6d522b : SUCCESS in 15m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c91d767b290f4caf831ea7fb9540f164 : SUCCESS in 16m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b9d431f9dfb44b9a22fbf015219d0d3 : SUCCESS in 40m 46s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6326f15419024ea096e7ffebb946e41e : SUCCESS in 1h 00m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1e672d36c7534f36873ac403c6859073 : SUCCESS in 21m 56s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a71a5a3c907e440f8c959a45019df519 : FAILURE in 9m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b134ccce668f49899bf782483845c75f : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b8bd882e1de801ec10e47483f645225cd3edd429","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-25 17:29:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2137d6da68ac1416218ebb55b71b9317a0b063c1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-25 19:02:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0fd57fb6f4938b00ecf7a52744983a3cd1d854b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:02:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c51a0bde8231ed81c4f06063229611e42e2925e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:10:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/136ee8a84227441cb18fe23ffe79a01c : SUCCESS in 5m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83bdda61597443a992e565f567bec04a : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3140b216a17141d4bc8f37dc9488f23a : SUCCESS in 5m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/244f5fc24fd9499f88e561d72ba08a6e : SUCCESS in 13m 17s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e2ac954cf854423b3b95c48a9d324c8 : SUCCESS in 11m 58s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b73d81ed3704308b63486cc907873d8 : SUCCESS in 1h 01m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9e93c2e6b2cf447fb2c674cef8ea3757 : SUCCESS in 20m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d5b36db9bb348998e28415ec0e03bf9 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"89d51e2fa1b1a726b5eb39b0b8b3830ae7695696","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:10:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9456349df064eb9deef37e97f333e798dd1a1f2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:12:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ec4eeca86ab475ea657b2308817eb9d : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1084af6e865f7722d0edc6fc8fe3864cf143f7df","revisions":{"1084af6e865f7722d0edc6fc8fe3864cf143f7df":{"kind":"REWORK","_number":1,"created":"2020-03-25 08:23:18.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/80/714880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/80/714880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/80/714880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/80/714880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/80/714880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/80/714880/1"}}},"commit":{"parents":[{"commit":"5a6838f7cd6bf44a5c4c0b5a88b1d7634a2affe9","subject":"Remove idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/5a6838f7cd6bf44a5c4c0b5a88b1d7634a2affe9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:41.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-25 06:53:41.000000000","tz":480},"subject":"Deprecate min_pool_size option","message":"Deprecate min_pool_size option\n\nmin_pool_size option is not used，see:\n\nhttps://review.opendev.org/#/c/565090/\n\nChange-Id: Ibd40fed67872c502f54f12c07ebe20fa47089aa5\nCloses-Bug: #1868511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1084af6e865f7722d0edc6fc8fe3864cf143f7df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1084af6e865f7722d0edc6fc8fe3864cf143f7df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
