)]}'
{"id":"openstack%2Fpuppet-heat~716451","triplet_id":"openstack%2Fpuppet-heat~stable%2Fstein~Ia251802612e7a4cf5c6828f3b7c1a0ef63a3c069","project":"openstack/puppet-heat","branch":"stable/stein","hashtags":[],"change_id":"Ia251802612e7a4cf5c6828f3b7c1a0ef63a3c069","subject":"Add server_keystone_endpoint_type param to heat::engine","status":"MERGED","created":"2020-03-31 23:14:13.000000000","updated":"2020-04-03 15:45:32.000000000","submitted":"2020-04-03 15:43:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716451-1585928628947-87f092a5","meta_rev_id":"f11cb4fecbb38f85a7bf6c27acb8abe9f67d8ad5","_number":716451,"virtual_id_number":716451,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 15:43:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-03 14:21:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-04-02 12:04:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-02 16:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-03 14:21:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 12:04:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-02 16:37:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-03 14:21:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-03 15:43:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14eb9bd06c6b0d74b40f747cb6a8992fb28e3c38","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-31 23:14:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28160e150568b2493f70f31a3a9858ed4a4eb967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 01:34:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ff7af0a6a4640e781476bf83bffebdb : SUCCESS in 5m 43s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8e7f6e9e28143d0868ea6e6c4ca0e7e : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acd77cdadc864a0c8ecce069baa6f598 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d3f520f98e14562a54fb059e1781fb4 : SUCCESS in 4m 59s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/3e9032cf4d7c458d9da438a1d686ff0a : SUCCESS in 18m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/171a70c4dc32416f845d25953180da33 : SUCCESS in 13m 18s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a70ef0b9b4f4b04a046fe9e030160cd : SUCCESS in 14m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cf9dd9c4bcf4e59b1a826e8e459b7e2 : FAILURE in 10m 48s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35b1a992fd3c46d8bb4a3fa2a4ab1a33 : FAILURE in 20m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a6a05e7f2c3d46d88a02167733ceebf7 : SUCCESS in 1h 01m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/680bf560ce8c4aef9b8e8dd323750fe5 : SUCCESS in 16m 24s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b76d07d17869421fbd650d95ed34d32d : FAILURE in 7m 29s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ee1f66e191ce413db6f461a6666a11d2 : SUCCESS in 2h 18m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25348ee8ce6044e594ed5027098451af : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b3caf11c93bba669b4536885958fbcc35d1a8e8f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-02 12:04:52.000000000","message":"Patch Set 1: Code-Review+1\n\nClean backport without affecting default. I confirmed the parameter exists in heat stable/train and stable/stein .","accounts_in_message":[],"_revision_number":1},{"id":"0ea9bf59e888725ba0542773b7e4b46ff037419d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-02 16:37:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"888c6eeab1a52c890bd83c3f16e306a9fd3d83da","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-03 14:21:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbd0e8ccc74303e230be763633a0b17acd6209f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:21:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"563370ebc53a3cedbfc11414fdf530deab54144a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:43:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b627a426aeb4fc291b2a7345d5950d9 : SUCCESS in 7m 14s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d61bfe6d63e450992c031be2a51cce2 : SUCCESS in 6m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61e738a3f4184c1ab16d8e6c321ea466 : SUCCESS in 6m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8128f5b20e6f4fcb871a17781b4771c8 : SUCCESS in 7m 16s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/4fcc88bfc9154323b6ef77d1f7e52bbe : SUCCESS in 26m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/55043a880c7d4eae875fc3a950e77e6b : SUCCESS in 25m 24s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65e9306c5d7948ffb81b55557501a2d5 : SUCCESS in 13m 15s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d25ba03196e549a893100dec002a9585 : SUCCESS in 1h 11m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d5f383f167c54f22b040258c7d85ffa6 : SUCCESS in 20m 48s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a8e5875caf7487b8c6678524f3c9561 : FAILURE in 8m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0ceee408cb742c38dcec4f0e09f5bb4 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"bfaab457babba569703fe690e1b54211b6e4c570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:43:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f11cb4fecbb38f85a7bf6c27acb8abe9f67d8ad5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:45:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afd57f113f514ef0a75add183b529887 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60f84ed59c0e6d43e2daad8706f87f7c1d7b3af7","revisions":{"60f84ed59c0e6d43e2daad8706f87f7c1d7b3af7":{"kind":"REWORK","_number":1,"created":"2020-03-31 23:14:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/51/716451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/51/716451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/51/716451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/51/716451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/51/716451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/51/716451/1"}}},"commit":{"parents":[{"commit":"4425351a7c9b357f5e80eea54540483c8ef7d094","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4425351a7c9b357f5e80eea54540483c8ef7d094"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-25 17:45:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-31 23:14:01.000000000","tz":120},"subject":"Add server_keystone_endpoint_type param to heat::engine","message":"Add server_keystone_endpoint_type param to heat::engine\n\nAdds the server_keystone_endpoint_type config option [1].\n\nThis is available all the way back from Stein and will\ntherefore be backported.\n\n[1] https://github.com/openstack/heat/blob/master/heat/common/config.py#L99\n\nChange-Id: Ia251802612e7a4cf5c6828f3b7c1a0ef63a3c069\n(cherry picked from commit 093b663165e9fc698b5c773343a3110980e88fe3)\n(cherry picked from commit efec41946c5e45926f43aa606e011eb5466688e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/60f84ed59c0e6d43e2daad8706f87f7c1d7b3af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/60f84ed59c0e6d43e2daad8706f87f7c1d7b3af7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
