)]}'
{"id":"openstack%2Fpuppet-heat~719540","triplet_id":"openstack%2Fpuppet-heat~master~I07b1b879a26364de0ea1a9fb15bc2fed1691bff6","project":"openstack/puppet-heat","branch":"master","topic":"prep-release-ussuri-3.1","hashtags":[],"change_id":"I07b1b879a26364de0ea1a9fb15bc2fed1691bff6","subject":"Prepare Ussuri M3 again","status":"MERGED","created":"2020-04-13 15:00:42.000000000","updated":"2020-04-13 15:15:37.000000000","submitted":"2020-04-13 15:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719540-1586790937846-61dee02a","meta_rev_id":"1ed301854f68b48ff9fd70fd7f7691f367f7bb3e","_number":719540,"virtual_id_number":719540,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 15:15:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-13 15:02:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-13 15:02:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 15:02:56.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-13 15:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4df4b0b7ac992134f4a4bf4a8e49060cc2fbc964","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-13 15:00:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa98e3a084715c641a7aa8bb0cb61796e1328b6e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-13 15:02:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb3d968b8ceab880fb0210d842246dd26668e3e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:06:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/4d364cf542ab49f4b049e25d68ca38e7 : SUCCESS in 3m 33s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b294e0b90854d2ba160b04086a38423 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"62a61ac8d7b0dce1e9dcf07908e8e65b38e2d315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:07:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2093b70da444e80eeac4d6cc61b7f3c978b544a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:15:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f9553837689f23772aaa69eb9d3ab6db7c2c5bf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:15:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/09eda384f7554adf93e4f378fc6f3be8 : SUCCESS in 5m 21s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb3d971368c94c39a43a45f8fff7a0d1 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d153dbea0e3f1ae1d859fed0c291c2aa4f8cb4bc","revisions":{"d153dbea0e3f1ae1d859fed0c291c2aa4f8cb4bc":{"kind":"REWORK","_number":1,"created":"2020-04-13 15:00:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/719540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/40/719540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/40/719540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/40/719540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/40/719540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/40/719540/1"}}},"commit":{"parents":[{"commit":"0b89f6ae3f7bc7461e7219ccbc31687bb0b7d1de","subject":"Resolve duplicated requirement names in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0b89f6ae3f7bc7461e7219ccbc31687bb0b7d1de"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-13 15:00:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-13 15:00:29.000000000","tz":540},"subject":"Prepare Ussuri M3 again","message":"Prepare Ussuri M3 again\n\nUpdate the version metadata to recreate release for Ussuri milestone M3\n\nChange-Id: I07b1b879a26364de0ea1a9fb15bc2fed1691bff6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/d153dbea0e3f1ae1d859fed0c291c2aa4f8cb4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/d153dbea0e3f1ae1d859fed0c291c2aa4f8cb4bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
