)]}'
{"id":"openstack%2Fpuppet-heat~729207","triplet_id":"openstack%2Fpuppet-heat~master~I1e0061aea6782bd61e745bb39683589794231a5f","project":"openstack/puppet-heat","branch":"master","topic":"db-password","hashtags":[],"change_id":"I1e0061aea6782bd61e745bb39683589794231a5f","subject":"Remove password hash generation in each puppet modules","status":"MERGED","created":"2020-05-19 11:00:19.000000000","updated":"2020-05-20 04:57:39.000000000","submitted":"2020-05-20 04:57:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729207-1589950659271-28dcb43a","meta_rev_id":"cd9cc48b6256deabe147ebc37b4438e2ffef5e6a","_number":729207,"virtual_id_number":729207,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-20 02:34:09.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 04:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-19 12:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-05-19 21:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-05-19 21:46:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 12:16:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-19 21:46:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-20 04:57:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d540b09c01c0677931dfe318d182a138766c8971","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 11:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f636378b5330702761df2a5295e6bb0bcb0b7eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:13:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2f054ad97c3f4f65821e9c74a9f1ae2a : SUCCESS in 2m 43s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f87498d97e614d7b9531392a623663da : SUCCESS in 4m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4abea07b795745fbbb81e7acc1ec8010 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/041f3cd44778434ab22ce229368a814b : SUCCESS in 5m 45s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3a092887d8b24d889f18d705d2b4adb7 : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cf8d4178d60145349898f450ef0782bd : SUCCESS in 13m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9f1f82291f3417bb26f67aac3e50b90 : SUCCESS in 10m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b6c7aaca6ce41129b1ecf01f63e749e : SUCCESS in 10m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8605635230d49e2b58cac6a8b32f57b : FAILURE in 32m 46s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4ff18c5f18474ab69fc0515cbf6caf8f : SUCCESS in 1h 11m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/13361eb119294710b3c69bfda0cffecd : SUCCESS in 18m 50s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6a08b2a2a9e54f978eb110f9cd5c73fc : FAILURE in 10m 50s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a86e5da8a0c46cf8a65f5e7650fa275 : FAILURE in 7m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bf6445d1fb51b5409e344630f9ff5a49590bfc4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-19 12:16:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"756e8a9a64ad1364505bbab0513d1b0330cad696","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-19 21:46:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"878016f7ab436f1e76cfa2fb69b2fa999da9f5bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-20 02:34:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a45812837716a06c6d31a5c997440799493707f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:42:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/39206c70fb7c41398212658fc3c7c176 : SUCCESS in 2m 59s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d558c09e5f534afd9090ff76b188552a : SUCCESS in 5m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cbadbb117074421b5bdc88f2b08ab79 : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6359b28b3191426d965e76cb0aea6fdb : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b8cdc0a9cb0343878e88b11b30236958 : SUCCESS in 14m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/62ffd33025504496a085ca8c33262b97 : SUCCESS in 18m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8973f1a8caa438d8b9ab1c5f6169e67 : SUCCESS in 12m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69bba20a170d4377a3fc65f4d3d26585 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f5a83147bc544f985d3622f2a18834d : FAILURE in 35m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cf31e590fd0444c1b0222ff150d1d575 : SUCCESS in 1h 03m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d4154887f87b421bb63ca6b2af559b42 : SUCCESS in 22m 38s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/fa7d57cc851c433f862fc055efba2c7a : FAILURE in 10m 56s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7208e894aca64486b53078ad0e2e0155 : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c53a47482a8bc024803f91179eca3de05b9db11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 03:42:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0006af01d3bf1152c348eebe2d653dbb2c386aee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:57:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ec813b563c444c96a6f932960c90ed81 : SUCCESS in 2m 38s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cac362ecf07c482eb91a6b01ec33a189 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c755f3e18b74b6da4d68f047a8549e1 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64b3327003ad47be8795c7ae6eea002f : SUCCESS in 4m 50s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/71e559cda4274f9eb738b923e5f90181 : SUCCESS in 14m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/536955580060401d90e2430603716e3e : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06a03f672d2244529541c77dbddc2321 : SUCCESS in 10m 49s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f794b04b2831499c98ea1b735853f77f : SUCCESS in 1h 09m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/508b0a88e0b74d6291cdfa47266744d3 : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2815919690d0957b5ce4a297355b2ac6e4845f50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 04:57:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1de8cdd7a43ac8b5a15460ebd6c99f636b901f7f","revisions":{"1de8cdd7a43ac8b5a15460ebd6c99f636b901f7f":{"kind":"REWORK","_number":1,"created":"2020-05-19 11:00:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/729207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/07/729207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/07/729207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/07/729207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/07/729207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/07/729207/1"}}},"commit":{"parents":[{"commit":"369d53129aba4f8037fa0ace0582eb7af556732a","subject":"Bump reno to 2.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/369d53129aba4f8037fa0ace0582eb7af556732a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:00:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 11:00:03.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: I1e0061aea6782bd61e745bb39683589794231a5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1de8cdd7a43ac8b5a15460ebd6c99f636b901f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1de8cdd7a43ac8b5a15460ebd6c99f636b901f7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
