)]}'
{"id":"openstack%2Fpuppet-heat~747944","triplet_id":"openstack%2Fpuppet-heat~stable%2Fussuri~I5d82fb7cd41d2fec0c31107ac9b2f1a89c98fe19","project":"openstack/puppet-heat","branch":"stable/ussuri","topic":"centos8-beaker-stable/ussuri","hashtags":[],"change_id":"I5d82fb7cd41d2fec0c31107ac9b2f1a89c98fe19","subject":"Use common class to manage apache service in beaker job","status":"MERGED","created":"2020-08-25 13:32:46.000000000","updated":"2020-08-26 08:25:28.000000000","submitted":"2020-08-26 08:25:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747944-1598430328400-69a35627","meta_rev_id":"1cd8212244fa61f6167ab2e6cfbdc7d929bbea74","_number":747944,"virtual_id_number":747944,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-08-26 07:24:43.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 08:25:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-26 02:13:44.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-08-25 20:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-26 02:13:44.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 20:08:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-26 02:13:44.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-26 08:25:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f21726075fc724d46c3c76b2e2467228f17d5e2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-25 13:32:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"391ba30127e688c1a8b7ae70da4de5f4fd4769c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 14:04:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15a2df8b026345a38b1324ed875ffb89 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/427a1180f0f8405faeaf495bdb5bcac9 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/653f010442544bb3b94912dfaf5b5ab4 : SUCCESS in 4m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d9a7121a0dc24e7d9c87c899f3e57cf4 : SUCCESS in 20m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/71c54599a3bd4308a6fb45fbaef4389f : FAILURE in 7m 57s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38c49ba24c8246bb9f1e1bda2cfee076 : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a07db2b6810b7f3b3c04dc05aa3db7b89db87ca","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-25 20:08:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a14b0867526deb2a7683b8a9bc7fffa334e0a849","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 02:13:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5ad7e464882540d81a963a4adb2b766598d999c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-26 07:24:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a63e569d9686b8c0557a1c1ccac7b91978708f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:58:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e1cd1e1f3c14c028733e74e156aa47e : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bc7c5eb3325480ba25f0519b46b6dd5 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b682356489d42489df1fb04b2541b25 : SUCCESS in 4m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c8239b30283547a79ea60e047006b654 : SUCCESS in 20m 17s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4ad0a84a78bd424d9f19d82b6e0f9d7e : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f39fbb83834d4e86bffd2d9d9cc210ba : FAILURE in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8edf13709790a42f8ddd594f4e489aa4efa835af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:59:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"307de6a846226c4a7741e0e802f4cc266f516dbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:25:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"67436a8d80ab74e73f71b3f9233a0c9f1a6740b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:25:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cd06e332f654414b9f753090d65748a : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c94f673b90a4fff97d4de2d0dfb75e2 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e584ffe602f4ae8924a6b0c1c58d040 : SUCCESS in 4m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/33ef63e1a02341ed8246fd42941973a4 : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9314ef2c8f805ae233943a1886f04d5de5f67f8b","revisions":{"9314ef2c8f805ae233943a1886f04d5de5f67f8b":{"kind":"REWORK","_number":1,"created":"2020-08-25 13:32:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/747944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/44/747944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/44/747944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/44/747944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/44/747944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/44/747944/1"}}},"commit":{"parents":[{"commit":"8b91724b4004d8e975ee0acbcdedf957ceb0421b","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8b91724b4004d8e975ee0acbcdedf957ceb0421b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-07 00:22:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-25 13:32:46.000000000","tz":0},"subject":"Use common class to manage apache service in beaker job","message":"Use common class to manage apache service in beaker job\n\nDepends-on: https://review.opendev.org/#/c/747935/\nChange-Id: I5d82fb7cd41d2fec0c31107ac9b2f1a89c98fe19\n(cherry picked from commit 312998a6e12c415cc0ba791765a3200a258552be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/9314ef2c8f805ae233943a1886f04d5de5f67f8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/9314ef2c8f805ae233943a1886f04d5de5f67f8b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
