)]}'
{"id":"openstack%2Fpuppet-heat~748086","triplet_id":"openstack%2Fpuppet-heat~master~I6392637052f305f6c0b9381923d8e9a69774a51a","project":"openstack/puppet-heat","branch":"master","topic":"bug/1892952","hashtags":[],"change_id":"I6392637052f305f6c0b9381923d8e9a69774a51a","subject":"Add mysql_enable_ndb option","status":"MERGED","created":"2020-08-26 03:54:21.000000000","updated":"2020-08-26 17:28:55.000000000","submitted":"2020-08-26 17:27:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748086-1598462839337-1545f216","meta_rev_id":"b7cf2b99b025c8df9dcdb26ca51d2eedfb8e7491","_number":748086,"virtual_id_number":748086,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 17:27:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-26 12:47:42.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-26 12:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-08-26 15:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-26 12:47:42.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-08-26 15:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 12:41:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-26 12:47:42.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-26 15:46:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-26 17:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b77a3035a7ccafe9c191bcc711e8a808620b32c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:54:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e5a884d1cf163d240ed01d4679d722d260b9bd0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:57:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9a40d4ab517dc06104302efbedacb26dc106b035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 05:17:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ab0cc55cac14941b534f28f8ee5e5e8 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecc609edfe434e8aac5887edd7f92e58 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a8a5f3256374fc49d9e85b7410d1e79 : SUCCESS in 5m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/da232dcabee74decbd9d2b7e6e2d9a65 : SUCCESS in 12m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c02d113647764eb3b795854aaf8116dd : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0be4df7e0250431888a4654f9d7a7275 : SUCCESS in 11m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac3172a372fa41c2aab2927717b6bb59 : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fdc17d7b1c443e89e6c8927c3ccd765 : SUCCESS in 40m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d74b7055946c44dc98a8007865fc0f8f : SUCCESS in 1h 16m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4baa89afb1784c6ebc22e6761a8a8fe4 : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/218a83870c764aa997e777cfc7890ebb : FAILURE in 6m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e98a330979354f1780257e16964210a1 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c18298edb6ff5e6d3aba36b10d5a68bd6191adc1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-26 12:41:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2c490b4fdc7522346bbd293e2933c8a3fea5353f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 12:47:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e30a2e72e7b779fc638d3f2bf38eea25d7f9c66b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-26 15:46:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0eb8950555ff3b7acbe1e7b13c1f3fcc7749790f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 15:47:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e8a7557fd4bf23e259a94acd6080d1f4c16019b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:27:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6a7c9771b4e4a5c2cb8ad9f0cf336fe834690912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:27:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bce983cf291d41a0abcd9a48bd73b5ee : SUCCESS in 3m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1da65df1581648db9f809effea6f542c : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f6a7c9ca6db4a4388100b57381cca73 : SUCCESS in 3m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0c8d6646096c465f80938001e6d53e84 : SUCCESS in 10m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/93ab8c1b3a53495c8b7116080448014e : SUCCESS in 10m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a4ebe1ca36d42abaff351586d22b67e : SUCCESS in 7m 26s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e1a94f0bcb944b0baed90141e76c939f : SUCCESS in 1h 29m 20s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0105273b856e42daa991ea6570c2505f : SUCCESS in 25m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71f8ce2324234a5db3982a72f18bc6b0 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c7bc11ea4d517d9db75cd065c58e5050e80e6199","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:28:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1560eff1010b4336bac426715e111336 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9fd21cf7cf21f08d982b3987336765a33f240779","revisions":{"e12d78fb9b4872152c11a10138069f0861f088e5":{"kind":"REWORK","_number":1,"created":"2020-08-26 03:54:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/86/748086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/86/748086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/1"}}},"commit":{"parents":[{"commit":"66b976f16807dcde0c3c5584e6e3ea8081a4a43d","subject":"Merge \"Expose vhost_custom_fragment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/66b976f16807dcde0c3c5584e6e3ea8081a4a43d"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: I6392637052f305f6c0b9381923d8e9a69774a51a\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e12d78fb9b4872152c11a10138069f0861f088e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e12d78fb9b4872152c11a10138069f0861f088e5"}]},"branch":"refs/heads/master"},"9fd21cf7cf21f08d982b3987336765a33f240779":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-26 03:57:22.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/86/748086/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/86/748086/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/86/748086/2"}}},"commit":{"parents":[{"commit":"66b976f16807dcde0c3c5584e6e3ea8081a4a43d","subject":"Merge \"Expose vhost_custom_fragment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/66b976f16807dcde0c3c5584e6e3ea8081a4a43d"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:55:57.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: I6392637052f305f6c0b9381923d8e9a69774a51a\nDepends-On: https://review.opendev.org/#/c/748067\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/9fd21cf7cf21f08d982b3987336765a33f240779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/9fd21cf7cf21f08d982b3987336765a33f240779"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
