)]}'
{"id":"openstack%2Fpuppet-heat~748880","triplet_id":"openstack%2Fpuppet-heat~master~I362610e251fa85cc60a8b0494775ee9db975cc37","project":"openstack/puppet-heat","branch":"master","topic":"validate-db_conn","hashtags":[],"change_id":"I362610e251fa85cc60a8b0494775ee9db975cc37","subject":"Do not validate database_connection format","status":"MERGED","created":"2020-08-30 15:40:52.000000000","updated":"2020-08-31 15:30:39.000000000","submitted":"2020-08-31 15:30:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748880-1598887839765-9d651cf1","meta_rev_id":"e1bc606fbaadf438f19f87581146c5753f7d6525","_number":748880,"virtual_id_number":748880,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-08-30 22:52:08.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 15:30:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-31 13:30:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-08-31 07:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-31 13:30:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 07:48:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-31 13:30:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-31 15:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99ff642d9c08cfbff81066f865a323192c2158b6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-30 15:40:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c8dacebb1ab769a379a4354dd33bfd73d08bfa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 17:11:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87aa4ad54cd441a98fe40ce036222796 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0e60b15693446d2bbf88742498e03e2 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df434522885f42088f34215ae3732cf8 : SUCCESS in 2m 50s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9e7599ca90f04c189970949489fa4983 : SUCCESS in 10m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c050ac0b76914dc4bdca8f6faed50e9b : SUCCESS in 10m 11s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1b8e6bec4fa435f8b53ffaa42625fb8 : FAILURE in 16m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8162976ed64944a29aa884d9bceb7d94 : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/535915ed39ca4b65877ce5675d64817f : SUCCESS in 45m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/82c1b1f14453406bb8850205e4e42c04 : SUCCESS in 1h 29m 10s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/69800fda05934fdfb33ba6c38b24ca25 : SUCCESS in 18m 47s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ff5d0fd0bbb40f59696e7cd504df705 : SUCCESS in 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e712df234fd7a4c807fb184bf9a1c68994815da3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-30 22:52:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a31b0657230d5be30fed1179be04d1293e05bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 00:17:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d04ab458f604acbb327749333e3f96a : SUCCESS in 3m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfb533c761b3444ca1e53d6555929df2 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca80e90c52184f9bb98569a83adf5aaa : SUCCESS in 2m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/68a095b1c0794c17a9e7a86216fce433 : SUCCESS in 10m 45s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c7d20f1861fb49c4bc83f5c05bbd11f6 : SUCCESS in 15m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7d1ec3d675b49c3b8e84f7df926e4e2 : SUCCESS in 11m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40e8ee5739a04496867ad6e422a4856a : SUCCESS in 10m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed7f071f4c414dfda191dec6decf74b0 : SUCCESS in 45m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/688084c143944a77ba9e2f0cf9b9ddaa : SUCCESS in 1h 23m 47s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/dffadec52a7f462b91b5a031132f9f5d : SUCCESS in 24m 07s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75849ead367446c695da7f2fc445bce6 : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"959b12ad16dc7d43c1b4e1c1eb611f5ce2b412eb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 07:48:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81170b781f0f9ccbd55e75b4676a8a04750690a1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-31 13:30:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92768b4e1d7866f571948eae5ee395cf292742e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:30:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21b46b96b8f940a76ccb2a9d608aeb19631abac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:30:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5d26e6ac77ed57c814bcb08cdcd913073d1ccc92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:30:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f20c4f68ab284285a00861edefefd130 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f451bc20d1504d1cacadc42aad2bf50d : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efb4e5562cda448e8f765311749c1b0b : SUCCESS in 3m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/527ca1f798c94f94933c72b3d89b18d8 : SUCCESS in 14m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/71b972328cfd46c786dae8beb9a12d46 : SUCCESS in 11m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a642f436ffa5424ab5ee8d04b68f529c : SUCCESS in 9m 56s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2e49ca8fdf244eeb860426220ff97cb9 : SUCCESS in 1h 25m 37s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/5fd6301beee24367afb1635ee940735f : SUCCESS in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7ebf5ba5580ffaf63b7c8579c50afbafc3daaac","revisions":{"a7ebf5ba5580ffaf63b7c8579c50afbafc3daaac":{"kind":"REWORK","_number":1,"created":"2020-08-30 15:40:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/748880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/80/748880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/80/748880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/80/748880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/80/748880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/80/748880/1"}}},"commit":{"parents":[{"commit":"312998a6e12c415cc0ba791765a3200a258552be","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/312998a6e12c415cc0ba791765a3200a258552be"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-30 15:40:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-30 15:40:42.000000000","tz":540},"subject":"Do not validate database_connection format","message":"Do not validate database_connection format\n\nCurrently we validate database_connection in 2 layers, each puppet\nmodules and puppet-oslo, however this makes it difficult to maintain\nvalidation pattern because we always need to fix both.\nThis patch removes the validation from each puppet modules so that\nwe need to maitain only one place, puppet-oslo to update validation\nlogic.\n\nChange-Id: I362610e251fa85cc60a8b0494775ee9db975cc37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/a7ebf5ba5580ffaf63b7c8579c50afbafc3daaac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/a7ebf5ba5580ffaf63b7c8579c50afbafc3daaac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
