)]}'
{"id":"openstack%2Fpuppet-heat~758557","triplet_id":"openstack%2Fpuppet-heat~master~I2c3a5d3c005963fe490310f6b81619cc85b28cc9","project":"openstack/puppet-heat","branch":"master","topic":"at-params","hashtags":[],"change_id":"I2c3a5d3c005963fe490310f6b81619cc85b28cc9","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2020-10-16 12:39:16.000000000","updated":"2020-10-19 09:56:01.000000000","submitted":"2020-10-19 09:53:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758557-1603101221177-81340386","meta_rev_id":"9b109c048bb8eec09a3a2913e270be35044b9845","_number":758557,"virtual_id_number":758557,"owner":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 09:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":1,"date":"2020-10-18 13:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":2,"date":"2020-10-18 08:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-10-19 07:54:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-10-19 07:54:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 12:39:27.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-10-18 08:35:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-18 13:09:30.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-10-19 07:54:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-19 09:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"844151108268f656f4abe9d7b580daef8f5b9c41","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2020-10-16 12:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a11da055b544fcc76430b6ae3997f4cfd48ab11","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-10-16 12:39:27.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"0194c6a4a9ee1d4d2cab55cf121f795b58d7ed91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 14:26:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e71f040c4fe145f9a1ca50d20f23be42 : SUCCESS in 5m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8acfcf2844b3422f97de816b9913aeba : SUCCESS in 4m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b75a4408697a402a92b750d653570d0e : SUCCESS in 12m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c68a426ed772419281d23385d23556b1 : SUCCESS in 11m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1b0b75fb4cb4783b356a43bdcbc58db : SUCCESS in 11m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8676c308fb84d2f8be02af438715fad : FAILURE in 25m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/37d6a73b66fd424889261f61ae1089b1 : SUCCESS in 1h 29m 29s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8173940d9eef488d922a9f8d747bf305 : SUCCESS in 24m 16s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0eda92b53cff4e69a77a926dbbc90c98 : SUCCESS in 19m 21s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/53b0fa01005244f9bee7fc88f1044540 : SUCCESS in 1h 40m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1220ecb8a1dd4a19bb56bc427f602ea3 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ef691d30b5acb67a19e605752d9abd91f4f1150e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-18 07:45:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThe change looks almost good to me, but it would be nice if you\u0027ll fix a typo in test example.","accounts_in_message":[],"_revision_number":1},{"id":"92a23a0e817ad8814720d97dec1d41c61688e895","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2020-10-18 08:25:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5737e69801354faf62eacecc5d4a7887c486a4e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-18 08:35:47.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":2},{"id":"94e7d91cf9c800c38683e27932da59114f77d1c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 09:29:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cf8fba1c2714567a7871bea4d110a50 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d8bec214a34e9abb88ab8f7ce64be4 : SUCCESS in 4m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/517a7c2a141243b3ad1dddbb2e3a7336 : SUCCESS in 9m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b13877a36c6a4e2dbd8e783690a909d2 : SUCCESS in 8m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d13510f5eeb4bff982ea925ddcc60ea : SUCCESS in 9m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d38b031ee30c46938c9ca8d7b2a0549a : FAILURE in 22m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a4d1e429dbba43fe9585b8b41e4b4c36 : SUCCESS in 1h 01m 53s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/44a4fd072e724d28ad8f6eca24711004 : SUCCESS in 17m 02s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58a352d3984e48049365d9218cab9a48 : SUCCESS in 15m 10s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2ef63972294b40f99ace45abbcabdda8 : SUCCESS in 1h 01m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77a84b1d8a9a43cda69c559f52a52f8e : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ee21112e0a483c57b10f65ce118a05a79000c68b","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-10-18 13:09:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"92955b466514c327c0484b609934046e70618c0b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-19 03:42:48.000000000","message":"Topic changed from at_params to at-params","accounts_in_message":[],"_revision_number":2},{"id":"d1a1716440eb28669c0fcfd57f2cd044db270cca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-19 07:54:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a28d8a031d8555483d97d121b3d7a39bd256cf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 07:55:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4a7c6d7952f7a1947c6420fb72bbaf091c916e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 09:53:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06058f3f040c4b95b931ac8675fdfb78 : SUCCESS in 6m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1253fd1e9984321b58160537316375a : SUCCESS in 6m 35s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/87201d1ae79f4334a6536d72446ca36e : SUCCESS in 13m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b00cc6240cc14bc298e4591ec6e10745 : SUCCESS in 12m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2f513db1ad5c472b8b6969de777f5b6f : SUCCESS in 1h 53m 32s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e681f1819d754b03aec0391741a95538 : SUCCESS in 22m 26s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e1bac8595a04662bdd823124ecac045 : SUCCESS in 17m 27s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0b7987e0e9d841379d81db74ec93bea2 : SUCCESS in 1h 30m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfebe4e2a2aa4f52896711d8d6c0d4ed : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"325487a6ccc037664cf5916fe2401c95a5c5306e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 09:53:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9b109c048bb8eec09a3a2913e270be35044b9845","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 09:56:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61dceeb3a7bf494a8b684fd747243c1d : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"261f4b1cf6afcd03f829304066494ab04d206221","revisions":{"675e17885aac7d8833eb6c25c979be8b523361a2":{"kind":"REWORK","_number":1,"created":"2020-10-16 12:39:16.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/57/758557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/57/758557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/1"}}},"commit":{"parents":[{"commit":"1de5ac5846c4e66c233988bf2a2251eaf888ae88","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1de5ac5846c4e66c233988bf2a2251eaf888ae88"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2020-10-16 12:35:45.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2020-10-16 12:39:13.000000000","tz":600},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I2c3a5d3c005963fe490310f6b81619cc85b28cc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/675e17885aac7d8833eb6c25c979be8b523361a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/675e17885aac7d8833eb6c25c979be8b523361a2"}]},"branch":"refs/heads/master"},"261f4b1cf6afcd03f829304066494ab04d206221":{"kind":"REWORK","_number":2,"created":"2020-10-18 08:25:05.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/57/758557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/57/758557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/57/758557/2"}}},"commit":{"parents":[{"commit":"1de5ac5846c4e66c233988bf2a2251eaf888ae88","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1de5ac5846c4e66c233988bf2a2251eaf888ae88"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2020-10-16 12:35:45.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2020-10-18 08:23:41.000000000","tz":600},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I2c3a5d3c005963fe490310f6b81619cc85b28cc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/261f4b1cf6afcd03f829304066494ab04d206221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/261f4b1cf6afcd03f829304066494ab04d206221"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
