)]}'
{"id":"openstack%2Fpuppet-heat~760757","triplet_id":"openstack%2Fpuppet-heat~master~Ia08ce1ffa507572f6eff0ad24415802b0b535a13","project":"openstack/puppet-heat","branch":"master","topic":"bug/1902158","hashtags":[],"change_id":"Ia08ce1ffa507572f6eff0ad24415802b0b535a13","subject":"Deprecate allow_insecure_clients option","status":"MERGED","created":"2020-11-02 06:42:56.000000000","updated":"2020-11-03 12:02:12.000000000","submitted":"2020-11-03 12:00:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760757-1604404843869-4fbb7558","meta_rev_id":"19760ceb0457c0226c4a3fc6125f3d1a792098ac","_number":760757,"virtual_id_number":760757,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-03 02:01:10.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 12:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-11-03 06:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-11-03 09:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-11-03 09:15:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 06:37:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-11-03 09:15:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-11-03 12:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ae4c83b9f57dd3bbdf115eef0bc7978923d57e2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-02 06:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbd7adf0a0ea258f1b4ff1f7e78985d9ca447d04","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-02 07:25:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ebd194d3f5d0a252d4ad4fa7ec02ba9934ee519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 09:48:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b2d03db09094b8fab7b4079fbabd02d : SUCCESS in 3m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4812be81cc944c781c14018ae04c397 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/00bb3807d85f41d5b4b6609705748101 : SUCCESS in 11m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e1e63dcfaa84100967e47a4538f0b06 : SUCCESS in 8m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8ddd5c0700d4cc8b44093b51014ffe3 : SUCCESS in 8m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e14151a2eb24d3b9f3b91a4358729da : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6a4d34170bc8424a886ab76e5522aaba : SUCCESS in 1h 05m 16s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f670742108434b09a895c5c8232919b7 : FAILURE in 11m 10s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac425c19791940138814035e7c76e9cf : SUCCESS in 14m 23s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7263a233bdfc499db2f4f1b913d88f6f : SUCCESS in 2h 19m 34s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/25f38f3799624d4d8238287d42d6f58a : SUCCESS in 1h 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/461e6198a0d84fcc84829b8edcd73fc1 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"094dd63355d10972848915aed829d451f6c25482","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-03 02:01:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"449b2217b7bf7bde769ff8b02db3894f89873fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 05:28:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa790974330141d9b4c1bbb2082c953a : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cf767eb940344729381c1b6ae0dceff : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ff1641ffb73e477e948cbb717fcb3f6c : SUCCESS in 11m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7d5d2b522f44b658802946db8aaa9f5 : SUCCESS in 8m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/629447d848ed4cdcb65d670b04acdda9 : SUCCESS in 8m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0476309b06d94b72942e1b873d2e5cbb : FAILURE in 6m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d24aafb83f24f288daf2674677147ea : SUCCESS in 1h 41m 07s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0ac194c0979f4c389b90e1f98d3b9c8e : SUCCESS in 18m 50s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95f6c13dd69145acb54f29636360cb0e : SUCCESS in 12m 54s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f4f92167029245998b2293597b8630ad : SUCCESS in 3h 08m 02s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/52880d1f8230467fba8f592d51b9c6da : SUCCESS in 1h 37m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02722283fad94510b8aedd124854dd62 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2cd09dbb6b735fdc52c23068dc9dbb367057c48b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-03 06:37:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df909d11cdafc82dc2ec870cf0f2daa36b3ec222","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-11-03 09:15:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b2fcab4666b5886df0305d1eda1de11edf16e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:15:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84d56b35a41057e21ca726a0cbdda6debb6babf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:00:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36c4a5595707464dab38f61db5d572fb : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfa59c2b7a37450288f987b8ed1817cf : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/906ab4d12a594cd5863fddf7a684853c : SUCCESS in 13m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/173c39995ecf407fb4522d5c20f78227 : SUCCESS in 10m 58s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/436fc8ea43bd4ffb84d0f0e0d0a66cf7 : SUCCESS in 1h 27m 05s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8852bd02f32548b09ea6eb44cb538174 : SUCCESS in 19m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/492bece78ae74959ac8c410e2761ba90 : SUCCESS in 24m 02s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8c53b060061e4afa981d1b16e3bd87c5 : SUCCESS in 2h 35m 33s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c523b5bb881e43cfa2b86b68516e4f08 : SUCCESS in 1h 20m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7170d56c2ef44e2ca1ab84d4a68f0b72 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"98b8a2ac40297631a15441cbf24de379aaa22cab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:00:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b59ea13be0de0a46a5f5706cb06ad8fe5bd3bc98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:02:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f0bab3d12d94dc79e6c3213436307ef : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb28a6a9a4befc178d20abbca27d5eeb28746974","revisions":{"d0de2ddb0649ca1efef6b97be891f40881137980":{"kind":"REWORK","_number":1,"created":"2020-11-02 06:42:56.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/57/760757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/57/760757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/1"}}},"commit":{"parents":[{"commit":"851cfb4b691212db24477a421a286a87848b20dc","subject":"Fix outdated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/851cfb4b691212db24477a421a286a87848b20dc"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 06:42:03.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 06:42:03.000000000","tz":480},"subject":"Deprecate allow_insecure_clients option","message":"Deprecate allow_insecure_clients option\n\nThe allow_insecure_clients has been deprecated[1].\n\n[1]https://review.opendev.org/#/c/417629/\n\nChange-Id: Ia08ce1ffa507572f6eff0ad24415802b0b535a13\nCloses-Bug: #1902158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/d0de2ddb0649ca1efef6b97be891f40881137980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/d0de2ddb0649ca1efef6b97be891f40881137980"}]},"branch":"refs/heads/master"},"bb28a6a9a4befc178d20abbca27d5eeb28746974":{"kind":"REWORK","_number":2,"created":"2020-11-02 07:25:40.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/57/760757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/57/760757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/57/760757/2"}}},"commit":{"parents":[{"commit":"851cfb4b691212db24477a421a286a87848b20dc","subject":"Fix outdated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/851cfb4b691212db24477a421a286a87848b20dc"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 06:42:03.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 07:25:27.000000000","tz":480},"subject":"Deprecate allow_insecure_clients option","message":"Deprecate allow_insecure_clients option\n\nThe allow_insecure_clients has been deprecated[1].\n\n[1]https://review.opendev.org/#/c/417629/\n\nChange-Id: Ia08ce1ffa507572f6eff0ad24415802b0b535a13\nCloses-Bug: #1902158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/bb28a6a9a4befc178d20abbca27d5eeb28746974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/bb28a6a9a4befc178d20abbca27d5eeb28746974"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
