)]}'
{"id":"openstack%2Fpuppet-heat~764764","triplet_id":"openstack%2Fpuppet-heat~master~I92552d013db98b972124db901f3f35a326110846","project":"openstack/puppet-heat","branch":"master","topic":"memcached-cert","hashtags":[],"change_id":"I92552d013db98b972124db901f3f35a326110846","subject":"Add TLS options to oslo.cache","status":"MERGED","created":"2020-11-30 14:35:36.000000000","updated":"2021-02-10 17:51:19.000000000","submitted":"2021-02-10 17:49:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764764-memcached-cert","meta_rev_id":"2c5c0f974218453eb34fb6721ffe683855f6ec64","_number":764764,"virtual_id_number":764764,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 17:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-02-10 12:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-02-10 14:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-02-10 14:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 16:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 16:40:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"},{"updated":"2021-02-10 12:00:34.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-02-10 14:48:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"100416f049ba386287bd44c56953755ca01f2e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2020-11-30 14:35:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f160486568b050fec7f898628fef14504846c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 16:14:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ddbcc641912480b8f72f8c2818802b3 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2356712fe5484ba1a51118861166ec21 : SUCCESS in 9m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6ce4c14826f9487ab2c9d2eaf9ba628f : SUCCESS in 11m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c52ec72c351414fb3ab28d67f4bafc1 : SUCCESS in 7m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4f28726f86a41a6a0e2c137cec9c193 : SUCCESS in 8m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92757b00bae0468eaf375bad275c6f9e : SUCCESS in 55m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ad90a2c0d0a94f53980e0be47d8b24c3 : SUCCESS in 1h 28m 04s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/04ae3d8f11b44ca299e8ca27d7e05d5e : SUCCESS in 19m 55s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d713e5835ea4e5b86013755448840e5 : SUCCESS in 15m 29s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7d5ac3b5d6fc4c409babcdb81aeb6d69 : FAILURE in 14m 58s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"00cffb928de17435fdcaad28f005ed9f0fb142a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2020-12-02 11:28:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92edacc70e897064940ca49a18b9337a66bdb1ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 13:15:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f81a830206dd4ae6893d3309bec85ddb : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80ee839ee26a4ee3ab16645126083e52 : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f3985ee5f5124a0f98c7f7ed82fb106c : FAILURE in 14m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be2dd9255add48268c510fb95f3259b4 : FAILURE in 13m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abeca3759c164fc292af33f45901dd1b : FAILURE in 14m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecafe60ff648422e9ca0bd6c1db21a36 : FAILURE in 11m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea9a15957c4e48ec9bac45edf1a2b7bf : FAILURE in 8m 41s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/6a400fb72fa24682994463032a69498f : FAILURE in 7m 38s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b01c4200f082474a9ba352509ec0fc05 : FAILURE in 5m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ee66c98edffa4b85bd778910e3f4af4b : SUCCESS in 1h 43m 21s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/304ac24fbb804b59b7489ce891000a13 : FAILURE in 35m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b71a87181153f6d243b5932dec31c8b819c71ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-01-12 12:42:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6c9db52a7ca16e1a227349ba551786fff39f9ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 16:13:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d805daaf1384f2787c66a677248ad85 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71890d3ebfa449c6a26bcf84ebb057e3 : SUCCESS in 3m 49s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2993d1352fb54bc081b01b1de2bc083b : SUCCESS in 9m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a70573b40897459d885679a72dcef30f : SUCCESS in 7m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99707fdf6aa74bc4954d833648ee940a : SUCCESS in 9m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa6bd8a8d0f14626b7fdcef7411c34de : SUCCESS in 44m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a585bfc48676498e95de99db4ce9e05f : SUCCESS in 1h 25m 07s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/3d75f79e1c8043f381bf8af87f2550de : SUCCESS in 19m 46s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abc8c4f0ea064e84a9b11f7eb6ef34bf : SUCCESS in 16m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ed06f6ad42444ef2add83f5ffc1e9eea : TIMED_OUT in 3h 27m 37s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":3},{"id":"8f712ae78932def21eaa99db9543a147adbe777c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-12 16:40:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"44895d6cdbe4142a776745d6bed5c8af73763af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 19:51:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e3ebbfcfaf541f9a64a440d0ef85334 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59c7b58ea9db48bca39dbed42a00d11d : SUCCESS in 3m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ae54f46bd753482d909fe507cef321ab : SUCCESS in 13m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f82521cd460e4f8cb4846fade503f227 : SUCCESS in 7m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ccf3b0d5f434289a1993cbd114735b6 : SUCCESS in 7m 40s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c99615dd48ce4f62becef17ca5c35769 : SUCCESS in 51m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9470a2cb25b24479957a56941be83afe : FAILURE in 10m 10s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/cb0a4bbada30491798f18308578710e6 : SUCCESS in 18m 22s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e34d102602c1482286f19b76b40a7a7d : SUCCESS in 13m 52s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8d969e6c63cb47ec877486773dc34832 : SUCCESS in 2h 45m 57s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f993705af74e409f82931ceecea66748 : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":3},{"id":"7071d09d85dfc211510c0d7577e0a60f8504763c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-01 08:59:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0011554cc6cf6487ac397e79cb39340221a0a8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 09:57:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad12b2b9d6524599ac7d54bd537a4d98 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/700abd1822144217adc1c749f61b6894 : SUCCESS in 4m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6a98793af72545a69c233ebeec664eba : SUCCESS in 12m 16s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/661d192d6ca348a8a2b1cbf9292e2b79 : SUCCESS in 11m 55s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec21204768f14e69b713c361c8b3056e : SUCCESS in 7m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30c41241066a4089aa81eecb9364ae70 : SUCCESS in 7m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fabd64ed73324b01956aad5b90979752 : SUCCESS in 37m 15s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/86adab1422ca4e6cacaf400d9d21e64f : FAILURE in 10m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e0da9ca94d24752a31b24f433ebd985 : FAILURE in 10m 07s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/2ea6f9d4bde344b891dcfb35e37b2fe6 : FAILURE in 8m 34s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/320a27d1b4b1422ca0aaabd162b02259 : SUCCESS in 13m 07s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/297886c8c8bb48b4af386d92bd23f703 : FAILURE in 14m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":3},{"id":"8cef5788be6ac717c8cfb55dde7a6603bd2ff0d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-01 14:57:03.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"70440b234760642b0ecb13e5978301292c1ac694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 17:58:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f87dea7e5b349148626bea79c2d6e25 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9526c609dc3446ca10b4a87fcaa9d7d : SUCCESS in 3m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9bc98a649e4f4ee8a45894d3203fa1f0 : SUCCESS in 13m 31s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd15a4c683cc4e5780e8e7135e1c05d0 : SUCCESS in 14m 28s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd388c50d27645598f7c97c0c6e684f8 : SUCCESS in 9m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cca4aea58354cefbfe0b3d5c5c50a71 : SUCCESS in 7m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d384dbc94e954e99bbdbeaacd82c7a98 : SUCCESS in 52m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6fd1e818248b46eab6604ecf0dda6298 : SUCCESS in 1h 27m 05s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e72595fd0f24b92aa3025b97ba7190f : SUCCESS in 1h 53m 36s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1718480925f0454f99a4950e54ee2b6b : SUCCESS in 18m 54s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e47ae191367490887c3713539d4bcad : SUCCESS in 15m 03s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cbbc2f2b8e674e4a94cfb294014fc512 : SUCCESS in 2h 40m 49s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3ba914b8d4074769931933fa140131f3 : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":4},{"id":"f10143cf68d4b50ef5065aa601ccf33311e31943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-10 12:00:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"24c0c3d5ef6c4dfae43cd3e364ecd1b4adf8bb5f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-10 12:01:41.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a8c6008cc08246a92bdc765d8530b776c6d1fa12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 14:42:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d22156ba1e964e148b78179ec7fc93d4 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0094259299de4376875816ca842dbbb3 : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/24f0f1280fc84173b8446d283b5cff7b : SUCCESS in 10m 49s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d5d113f87c24c98bc21555539e8ebbf : SUCCESS in 10m 47s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ccb8e68dc0c4739bbad2b3ae809abc9 : SUCCESS in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83788cf3b27e4545a06e31ecf81136ae : SUCCESS in 8m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b2f04917a9d440ea5b8be4853bb98e3 : FAILURE in 34m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/429ab597ed53497d8c002ad7965de3d8 : SUCCESS in 1h 26m 17s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a0f5b72d5f245a9bb68f8eb2e5b9ff7 : SUCCESS in 1h 24m 43s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5394c8e682534d05aac3109e04a92068 : SUCCESS in 19m 04s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3e3614bcb8a49b3a70640b5c9a78bab : SUCCESS in 14m 44s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c98ea465b6ba43e1813fe9270980a529 : SUCCESS in 2h 34m 29s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ef2ecdfffb4f42948014793bb1c15a75 : SUCCESS in 1h 15m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/103c1674f5af4c72b9ed41e0bf20d19c : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":5},{"id":"1e6b0ffe2459a5c4d25a071195ab28ea67159b47","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-10 14:48:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ff30ccb72a9beafbcd907f4f9488c13af8984c76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 14:49:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4a110d5784563f6aa2f5eb542ec69ed4b6b01be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:49:25.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fb6a59b82db4d60890e140bc257bf3f : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e6cf71c626c4e2aa30ae7dce1c67a66 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fb5ca0007da5404b83304529ccd5edb0 : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96a9fa15c824881ac56aab09f974352 : SUCCESS in 10m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5571cfe7ecbe4cad9be2024d4629c0a6 : SUCCESS in 1h 26m 18s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/96d397fad226442b87d1d362bfb8edb3 : SUCCESS in 21m 05s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98604594abd043b69c2c3d8ba8a6bbf9 : SUCCESS in 14m 34s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/207a47f97d544e6aab697722de21ffc8 : SUCCESS in 2h 52m 14s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7d159b16355f4f839145f3fd6564767d : SUCCESS in 1h 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37b54f6deabc4c20ae1a1b1042a619eb : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":5},{"id":"96fbbed8861459b63ca34b11cad064d4b680eb40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:49:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2c5c0f974218453eb34fb6721ffe683855f6ec64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 17:51:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61772798097f427f91bd14d75f3ec824 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"42d1e8567d7441633ac9770c9301d8c5e5c0acd1","revisions":{"e7c1fba5ae64e697d1170d32d7e73e77fdaf4ade":{"kind":"REWORK","_number":1,"created":"2020-11-30 14:35:36.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/64/764764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/64/764764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/1"}}},"commit":{"parents":[{"commit":"c16009b55bb4e6a11c64e3efcae05aaeed453729","subject":"Allow db sync timeouts to be configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c16009b55bb4e6a11c64e3efcae05aaeed453729"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-30 14:34:43.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-30 14:35:32.000000000","tz":60},"subject":"[WiP] Add TLS options to oslo.cache","message":"[WiP] Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nDepends-on: https://review.opendev.org/761604\nChange-Id: I92552d013db98b972124db901f3f35a326110846\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e7c1fba5ae64e697d1170d32d7e73e77fdaf4ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e7c1fba5ae64e697d1170d32d7e73e77fdaf4ade"}]},"branch":"refs/heads/master"},"58fca454d51f0875b6e4929cf4858f40bee6d6cf":{"kind":"REWORK","_number":2,"created":"2020-12-02 11:28:30.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/64/764764/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/64/764764/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/2"}}},"commit":{"parents":[{"commit":"c16009b55bb4e6a11c64e3efcae05aaeed453729","subject":"Allow db sync timeouts to be configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c16009b55bb4e6a11c64e3efcae05aaeed453729"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-30 14:34:43.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-12-02 11:28:21.000000000","tz":60},"subject":"[WiP] Add TLS options to oslo.cache","message":"[WiP] Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nDepends-on: https://review.opendev.org/761604\nChange-Id: I92552d013db98b972124db901f3f35a326110846\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/58fca454d51f0875b6e4929cf4858f40bee6d6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/58fca454d51f0875b6e4929cf4858f40bee6d6cf"}]},"branch":"refs/heads/master"},"dca5b9765f5c93b93e256726c08029349dbbc9d5":{"kind":"REWORK","_number":3,"created":"2021-01-12 12:42:56.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/64/764764/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/64/764764/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/3"}}},"commit":{"parents":[{"commit":"c16009b55bb4e6a11c64e3efcae05aaeed453729","subject":"Allow db sync timeouts to be configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c16009b55bb4e6a11c64e3efcae05aaeed453729"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-30 14:34:43.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-01-12 12:42:15.000000000","tz":60},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nDepends-on: https://review.opendev.org/761604\nChange-Id: I92552d013db98b972124db901f3f35a326110846\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/dca5b9765f5c93b93e256726c08029349dbbc9d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/dca5b9765f5c93b93e256726c08029349dbbc9d5"}]},"branch":"refs/heads/master"},"4c8558a4182affaeda71d18b992ce99d0aa9d596":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-02-01 14:57:03.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/64/764764/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/64/764764/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/4"}}},"commit":{"parents":[{"commit":"c16009b55bb4e6a11c64e3efcae05aaeed453729","subject":"Allow db sync timeouts to be configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c16009b55bb4e6a11c64e3efcae05aaeed453729"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-30 14:34:43.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-02-01 14:57:03.000000000","tz":0},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nChange-Id: I92552d013db98b972124db901f3f35a326110846\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4c8558a4182affaeda71d18b992ce99d0aa9d596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4c8558a4182affaeda71d18b992ce99d0aa9d596"}]},"branch":"refs/heads/master","description":"Edit commit message"},"42d1e8567d7441633ac9770c9301d8c5e5c0acd1":{"kind":"REWORK","_number":5,"created":"2021-02-10 12:00:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/764764/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/64/764764/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/64/764764/5"}}},"commit":{"parents":[{"commit":"c16009b55bb4e6a11c64e3efcae05aaeed453729","subject":"Allow db sync timeouts to be configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c16009b55bb4e6a11c64e3efcae05aaeed453729"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2020-11-30 14:34:43.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-10 12:00:25.000000000","tz":540},"subject":"Add TLS options to oslo.cache","message":"Add TLS options to oslo.cache\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe oslo.cache backends that supports TLS connections.\n\nDepends-on: https://review.opendev.org/761604\nChange-Id: I92552d013db98b972124db901f3f35a326110846\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/42d1e8567d7441633ac9770c9301d8c5e5c0acd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/42d1e8567d7441633ac9770c9301d8c5e5c0acd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
