)]}'
{"id":"openstack%2Fpuppet-heat~770286","triplet_id":"openstack%2Fpuppet-heat~master~Ic445e6d1f50aadc26349c7cd25673a4a46f10145","project":"openstack/puppet-heat","branch":"master","topic":"memcache-authtoken","hashtags":[],"change_id":"Ic445e6d1f50aadc26349c7cd25673a4a46f10145","subject":"[WiP] Add TLS options to authtoken middleware","status":"ABANDONED","created":"2021-01-12 12:42:56.000000000","updated":"2022-07-23 13:50:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a908678bb428bfa4c4b8391a71bcb04eb36d887","_number":770286,"virtual_id_number":770286,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-12 15:39:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":-1,"date":"2021-01-12 12:43:34.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 15:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83abd905a79776dda23b3a24e5f6234eb04d4f35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-01-12 12:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a90fbf460acbbf5aa9db49097ad886fb10ecec7","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-01-12 12:43:34.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9cee3457044627293b45ca0a066d5fdd0cfaed9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:39:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/277547f066264e6cab83bb33f6e03004 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5832ee955e04451c9e3495c7417fc953 : SUCCESS in 6m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0cedea000b114d58af16fe6d0f013b7e : FAILURE in 15m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52fbc8f8ef8141688a2595c882ea159f : FAILURE in 11m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b27be4ce7dd943df95ab5f320b1c13ac : FAILURE in 11m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db34d81e79f3426dbae7f2ad7f85fd8b : FAILURE in 6m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6e88286e694e452ca0e63fc1fed0590c : FAILURE in 15m 51s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/da543871a935432f9596efe731645781 : FAILURE in 9m 32s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ff3c9d7844544fd9880ecdd24f0f4d8 : FAILURE in 6m 08s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6d5a5fe5361245e79b9eb342afd81654 : SUCCESS in 2h 04m 53s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c439aaaa49bf4e03adbe717e7c223bc4 : FAILURE in 45m 33s","accounts_in_message":[],"_revision_number":1},{"id":"008aeaba11f7211a8e465aef950beddcad35ce78","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-04 19:07:52.000000000","message":"Topic memcached-cert removed","accounts_in_message":[],"_revision_number":1},{"id":"01183cd4ce17f8bd1f17eb551def666f715f1f62","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-04 19:07:56.000000000","message":"Topic set to memcache-authtoken","accounts_in_message":[],"_revision_number":1},{"id":"8a908678bb428bfa4c4b8391a71bcb04eb36d887","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-23 13:50:40.000000000","message":"Abandoned\n\nThis patch has been inactive for more than 1 year. The dependent patch has not been merged. So I\u0027ll abandone this to clean up our review queue.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a0d48d52ab4441432b35867dabefcc2d13a6019","revisions":{"7a0d48d52ab4441432b35867dabefcc2d13a6019":{"kind":"REWORK","_number":1,"created":"2021-01-12 12:42:56.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/86/770286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/86/770286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/770286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/770286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/86/770286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/86/770286/1"}}},"commit":{"parents":[{"commit":"dca5b9765f5c93b93e256726c08029349dbbc9d5","subject":"Add TLS options to oslo.cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/dca5b9765f5c93b93e256726c08029349dbbc9d5"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-01-12 12:31:40.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-01-12 12:42:31.000000000","tz":60},"subject":"[WiP] Add TLS options to authtoken middleware","message":"[WiP] Add TLS options to authtoken middleware\n\nThis patch specifies a set of options required to build a TLS context.\nThe context built from those options can later on be passed to any of\nthe keystone authtoken cache backends that supports TLS connections.\n\nDepends-on: https://review.opendev.org/766735\n\nChange-Id: Ic445e6d1f50aadc26349c7cd25673a4a46f10145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7a0d48d52ab4441432b35867dabefcc2d13a6019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7a0d48d52ab4441432b35867dabefcc2d13a6019"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
