)]}'
{"id":"openstack%2Fpuppet-heat~785261","triplet_id":"openstack%2Fpuppet-heat~master~I0e226046f6e2d69c89681948cee8e8830c186489","project":"openstack/puppet-heat","branch":"master","topic":"debian-uwsgi-support","hashtags":[],"change_id":"I0e226046f6e2d69c89681948cee8e8830c186489","subject":"Add support for heat_api_{cfn_,}uwsgi_config in Debian","status":"MERGED","created":"2021-04-07 19:42:41.000000000","updated":"2021-04-14 17:20:09.000000000","submitted":"2021-04-14 17:18:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785261-debian-uwsgi-support","meta_rev_id":"a4e9a653d7b11574a01f42226952c73f95bce33c","_number":785261,"virtual_id_number":785261,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 17:18:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"all":[{"value":1,"date":"2021-04-12 06:50:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":2,"date":"2021-04-14 14:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-04-12 06:56:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"value":2,"date":"2021-04-12 16:50:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-12 06:55:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":1,"date":"2021-04-12 06:45:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":1,"date":"2021-04-14 14:15:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 21:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 06:45:53.000000000","updated_by":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"reviewer":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"state":"REVIEWER"},{"updated":"2021-04-12 06:50:28.000000000","updated_by":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"reviewer":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"state":"REVIEWER"},{"updated":"2021-04-12 06:55:00.000000000","updated_by":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"reviewer":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"state":"REVIEWER"},{"updated":"2021-04-12 06:56:56.000000000","updated_by":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"reviewer":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"state":"REVIEWER"},{"updated":"2021-04-12 07:12:30.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-04-12 07:12:30.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-04-12 07:12:30.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-12 07:12:30.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e0401f521f8b4d2c6acf156aa4a34e65203a7218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 19:42:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a505cd51c1b4d57290c59a573cd05ff354426f58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 21:14:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5357387646a49e49ca3b679da975a07 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cec24e1db9f64fb798726db2a93de9b3 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c08cce33921347e38ab93afa6ca225f7 : SUCCESS in 9m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/369f0a9180474ca7b780f1d2e0d99e2a : SUCCESS in 7m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87ca3cf509944a29bcd6c6c41e46195b : SUCCESS in 8m 54s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f950930f72a42ec8f5488213f61970b : FAILURE in 29m 43s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebdb4b09f83c4c17904065ca265ec4a0 : SUCCESS in 1h 20m 50s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a302460ac6d4eb1939056effd0f25fc : SUCCESS in 21m 14s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d6d63080c8040daa3efcbc6e5501f77 : SUCCESS in 12m 33s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5ba8a655af034fb1befc09ca33eafefa : SUCCESS in 1h 29m 37s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ee25f0a5371b4cdf9575c68499daa1df : FAILURE in 34m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e89e55ba246848e1ae45731bfbccdecf : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b9fe75db4debfa1810124ec72693b2856448876d","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 21:58:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e5396a672be41a99b1d4796152917fc5c48dbf5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:35:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3024a990bdab4b9a822966777d02ed96 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa0d80f106ce4a4ba271d13280e0e469 : SUCCESS in 4m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f0acf722b3a48eda62ddbbfe074ae26 : SUCCESS in 10m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20e8c4ffc8cc4aceba3e2ca0ed790fb4 : SUCCESS in 8m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ca58568d48b4666a7e8f1654fc2977b : SUCCESS in 8m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2e3cb0f1d0f40daa1d82f972c0890bd : FAILURE in 29m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/918c1a3b0d994166b36346921159d081 : SUCCESS in 1h 28m 39s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ba5fc103c784d9da8f5cc1e4287189b : SUCCESS in 18m 18s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/330f937d789e498491e6f49ae84a412b : SUCCESS in 17m 32s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/11a40c2543a949e6a572786fd8674e5a : SUCCESS in 1h 28m 13s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1acf3178a2c743fd87f0d804017aa052 : FAILURE in 32m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/019b47be995e48f0aaccd542aaf30c61 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ea29b89870a738137a43011e02780b0c3b8812f2","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-08 07:52:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"744c2968a187dee54ab480061c46b209d47b40d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:55:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7ae9d21b274490998b5aca2550e4247 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f532a04827074a1fb4fd863603bfb8de : SUCCESS in 4m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/53882e288df54890845861d3e1c32ce4 : FAILURE in 9m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fcec640d2c2489ea96e1e5f145f50d9 : FAILURE in 6m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db0c2d520fab49839dcbe16aafd005ed : FAILURE in 6m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bda81cd3d61448d9b35ff7e32c4d17ac : FAILURE in 10m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31d75bb44df84e7aa72fa8eddb1abfd7 : FAILURE in 10m 35s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c89a59352e34188b821f20c5b3c254d : FAILURE in 11m 26s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c3fc729476e437fbed77d7351ed158b : FAILURE in 7m 19s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/12075e1e0e2e4659b1a93ecfa00d2d52 : SUCCESS in 1h 57m 45s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c2cd710f51254b6ea55edbe227948984 : FAILURE in 42m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aa45590db464313b769d2b134caa2bd : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"33347dc2231991751190a72c87d41cd1d468716d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-09 21:45:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6256c81fa1f5f7a1666f875b5cc2c092fdb46bfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-09 23:01:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d25640dfe05d80772f413827472d72c1c9c6f713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 01:39:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b007d44c8ca489089f6745659c0b736 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c375819309b24262a6106b3c6d1d8ef6 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c86f6b4709a469191617cf9e9b061bc : FAILURE in 8m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4cdc735a9624cce8f96faeb7a331849 : FAILURE in 5m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f56f6fe9f315495c8a88be10c8b25162 : FAILURE in 6m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0327cc1b85c44bc190b25d227cc4f7f3 : FAILURE in 28m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ac7660ab0b1486cb2ade3456d8c6c5c : SUCCESS in 1h 25m 09s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/64334d73ae0049eaa102274bf6e0b2ba : SUCCESS in 21m 09s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07e6599091f4aa69b4be66f39a4191a : FAILURE in 14m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5498a9a50c6e446994635c938d33bcd7 : SUCCESS in 2h 34m 16s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/99823643171e4770af736a047a4d8a41 : SUCCESS in 1h 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ce590660c334517b5ef629645eb7fbf : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"85a1ce2645270e2fb951e71a0a5469a3a40f41b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-10 16:27:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bf9ed2021febaec3a957ccf4163302b40a75c269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 18:12:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98974273365e44629355facda15512a4 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/586eb05744214498888a155b5a7d4ac4 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1509ab75c454eff9079d1ad3f800363 : FAILURE in 6m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e53f3e434d44291bf2edbd61a476ceb : FAILURE in 6m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c97319fbb624ca2ad831da5ac178b11 : FAILURE in 6m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cd35d6682534b7a8075f0f0a13e416b : FAILURE in 29m 16s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c32665960b0444539076646588080d4e : SUCCESS in 1h 04m 15s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d11aea250ff446492d74b0d53d48f7c : SUCCESS in 14m 34s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02c99d9f23064d04a4b649065ede599e : FAILURE in 15m 09s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3e6d8978d74e408ca1cdea13cb70a711 : SUCCESS in 1h 44m 08s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b3ad840ef5844cb5a83a27cd919ef182 : SUCCESS in 54m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0da7a1d245f444c8a021201202c263b : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"220e2cac8ce57d338feb9b1c3c2b65fa8e023c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-10 20:44:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"946e578d39eb70a71c97bf015810f3b9e06ea5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-11 00:03:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8630ac705fa944439aea50e468c6ac3a : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/899a8b6f49a546878b38a483e22e4b6e : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/12f8fbcfe4da44a7967d240d455ac9d7 : SUCCESS in 12m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/accf3b0fdd3b4acfaf30430be12f4cf9 : SUCCESS in 7m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1259384a3f07432491069be7e1f0cfde : SUCCESS in 7m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef0f04f9b0df472db91bb7044aab4fc0 : FAILURE in 28m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f28d18e00c8443509beb1f4b0a625bb4 : SUCCESS in 58m 02s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a32a6580365148eb96dce8b49376e76d : SUCCESS in 19m 06s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5a78a724d7543728ac540089c68106d : SUCCESS in 15m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/90e57a120ed444fe99fc1c2fdaf7ce4a : SUCCESS in 3h 11m 51s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/dc834657f30c4c40840b18707c32ea5c : SUCCESS in 1h 36m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c89beddd936340db86968d515ce361ac : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":5},{"id":"db716ba5cbbeaafa0f816c43d5610b716f9b1dcb","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2021-04-12 06:45:53.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"57adccfdbfef74983611c9a5060a770dbb1c0c78","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2021-04-12 06:50:28.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fb42a0bf5976ab918fd9612c46d149588c30d23b","author":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"date":"2021-04-12 06:55:00.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2b50f86b30c8f39dc55e1beabecc1f96fddded15","author":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"date":"2021-04-12 06:56:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2c0617c5f9f4f3cf16f91983b9d87b450c8ce63d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-12 16:50:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fa79e01c966518a3e6ef5ed2b457e70f5240fd76","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-14 14:15:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7ca8b8b671b2ee7baaae9a57c8e87aa2864cb073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:15:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e729e10999df3affdaa4e411d9ba4643624c1b84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:18:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dfdd1e7c1bd41ad8e6f869ade641055 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/516c8303d9fe4264ae1bbb37be3b91bd : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f10729f25804daebcc09326e8b5a45e : SUCCESS in 9m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b277c2b3a8a4d5ab4351679306f6fdd : SUCCESS in 7m 49s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2cb3fa263ab43a9993d5a730c686cbd : SUCCESS in 1h 07m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbabea923bf14756a233a4c26879659e : SUCCESS in 17m 11s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/941fa5841ed94424a1be557eed6db1da : SUCCESS in 12m 44s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8f15daf3889d4c97b38989b34ba8eb8d : SUCCESS in 1h 11m 16s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5967944130394e2482438f73e0ecb431 : SUCCESS in 1h 29m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e01c8d416c45402fb6084945f49b3867 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":5},{"id":"6713545f3ec753748a903ed6981b577c0328a32c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:18:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a4e9a653d7b11574a01f42226952c73f95bce33c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:20:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89b210d714b84c5a8aad16fa2064a2bc : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"274e458478225b66e4c4fba600b5aabd4f354553","revisions":{"9ba2857466e301c62dc2c7348c60aafeddd8a21e":{"kind":"REWORK","_number":1,"created":"2021-04-07 19:42:41.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/61/785261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/61/785261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/1"}}},"commit":{"parents":[{"commit":"644eeb990aa4e435ed88a71bf96f97c4eb004abb","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/644eeb990aa4e435ed88a71bf96f97c4eb004abb"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 19:37:14.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 19:42:06.000000000","tz":120},"subject":"Add support for heat_api_{cfn_,}uwsgi_config in Debian","message":"Add support for heat_api_{cfn_,}uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Heat API and api-metadata. Therefore, this patch adds\ntwo new heat_api_uwsgi_config and heat_api_cfn_uwsgi_config\nproviders as well as two new heat::wsgi::uwsgi_api and\nheat::wsgi::uwsgi_api_cfn classes.\n\nChange-Id: I0e226046f6e2d69c89681948cee8e8830c186489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/9ba2857466e301c62dc2c7348c60aafeddd8a21e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/9ba2857466e301c62dc2c7348c60aafeddd8a21e"}]},"branch":"refs/heads/master"},"28d7abd1e0a23a840e996d48024d76698c25ea18":{"kind":"REWORK","_number":2,"created":"2021-04-09 21:45:53.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/61/785261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/61/785261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/2"}}},"commit":{"parents":[{"commit":"644eeb990aa4e435ed88a71bf96f97c4eb004abb","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/644eeb990aa4e435ed88a71bf96f97c4eb004abb"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 19:37:14.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-09 21:45:40.000000000","tz":120},"subject":"Add support for heat_api_{cfn_,}uwsgi_config in Debian","message":"Add support for heat_api_{cfn_,}uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Heat API and api-metadata. Therefore, this patch adds\ntwo new heat_api_uwsgi_config and heat_api_cfn_uwsgi_config\nproviders as well as two new heat::wsgi::uwsgi_api and\nheat::wsgi::uwsgi_api_cfn classes.\n\nChange-Id: I0e226046f6e2d69c89681948cee8e8830c186489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/28d7abd1e0a23a840e996d48024d76698c25ea18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/28d7abd1e0a23a840e996d48024d76698c25ea18"}]},"branch":"refs/heads/master"},"e38d7b51e55fccdbcb4d2a0a4b6968598a976599":{"kind":"REWORK","_number":3,"created":"2021-04-09 23:01:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/61/785261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/61/785261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/3"}}},"commit":{"parents":[{"commit":"644eeb990aa4e435ed88a71bf96f97c4eb004abb","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/644eeb990aa4e435ed88a71bf96f97c4eb004abb"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 19:37:14.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-09 23:01:12.000000000","tz":120},"subject":"Add support for heat_api_{cfn_,}uwsgi_config in Debian","message":"Add support for heat_api_{cfn_,}uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Heat API and api-metadata. Therefore, this patch adds\ntwo new heat_api_uwsgi_config and heat_api_cfn_uwsgi_config\nproviders as well as two new heat::wsgi::uwsgi_api and\nheat::wsgi::uwsgi_api_cfn classes.\n\nChange-Id: I0e226046f6e2d69c89681948cee8e8830c186489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e38d7b51e55fccdbcb4d2a0a4b6968598a976599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e38d7b51e55fccdbcb4d2a0a4b6968598a976599"}]},"branch":"refs/heads/master"},"8e3a4fb7fbacee4ccac645e36eff3a2f74145b2e":{"kind":"REWORK","_number":4,"created":"2021-04-10 16:27:31.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/61/785261/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/61/785261/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/4"}}},"commit":{"parents":[{"commit":"644eeb990aa4e435ed88a71bf96f97c4eb004abb","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/644eeb990aa4e435ed88a71bf96f97c4eb004abb"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 19:37:14.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-10 16:27:18.000000000","tz":120},"subject":"Add support for heat_api_{cfn_,}uwsgi_config in Debian","message":"Add support for heat_api_{cfn_,}uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Heat API and api-metadata. Therefore, this patch adds\ntwo new heat_api_uwsgi_config and heat_api_cfn_uwsgi_config\nproviders as well as two new heat::wsgi::uwsgi_api and\nheat::wsgi::uwsgi_api_cfn classes.\n\nChange-Id: I0e226046f6e2d69c89681948cee8e8830c186489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8e3a4fb7fbacee4ccac645e36eff3a2f74145b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8e3a4fb7fbacee4ccac645e36eff3a2f74145b2e"}]},"branch":"refs/heads/master"},"274e458478225b66e4c4fba600b5aabd4f354553":{"kind":"REWORK","_number":5,"created":"2021-04-10 20:44:46.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/61/785261/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/61/785261/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/61/785261/5"}}},"commit":{"parents":[{"commit":"644eeb990aa4e435ed88a71bf96f97c4eb004abb","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/644eeb990aa4e435ed88a71bf96f97c4eb004abb"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 19:37:14.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-10 20:44:24.000000000","tz":120},"subject":"Add support for heat_api_{cfn_,}uwsgi_config in Debian","message":"Add support for heat_api_{cfn_,}uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Heat API and api-metadata. Therefore, this patch adds\ntwo new heat_api_uwsgi_config and heat_api_cfn_uwsgi_config\nproviders as well as two new heat::wsgi::uwsgi_api and\nheat::wsgi::uwsgi_api_cfn classes.\n\nChange-Id: I0e226046f6e2d69c89681948cee8e8830c186489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/274e458478225b66e4c4fba600b5aabd4f354553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/274e458478225b66e4c4fba600b5aabd4f354553"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
