)]}'
{"id":"openstack%2Fpuppet-heat~793520","triplet_id":"openstack%2Fpuppet-heat~stable%2Frocky~I4ff93cf4a1f225f88750fd37a392d36cb4dce7bb","project":"openstack/puppet-heat","branch":"stable/rocky","topic":"rocky-tripleo","hashtags":["puppet-tripleo-job"],"change_id":"I4ff93cf4a1f225f88750fd37a392d36cb4dce7bb","subject":"Rocky only: Remove TripleO Job","status":"MERGED","created":"2021-05-28 01:12:57.000000000","updated":"2021-06-11 10:10:29.000000000","submitted":"2021-06-11 10:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793520-rocky-tripleo","meta_rev_id":"f19ae6aed96c12667ed6fce479ab320b4aa9c28e","_number":793520,"virtual_id_number":793520,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 10:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-06-11 09:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-06-02 15:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2021-06-11 09:06:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 02:06:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 15:45:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-11 09:06:48.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"ddf46b60d5b0da9d9b3c4e50842f172d7d58abad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-28 01:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43824290cdc5324e814587a9e5c58acba3b9a344","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-28 01:39:12.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"8e6e70dfa4543b7baf9fc74d3bc01c146b688912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 02:06:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/bf21c2e12fdb4a2488f550b748fec849 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/0d936f0ad79a4e0d8b965464806077c2 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/567a53d16c27447fa55fa2358f34aba2 : SUCCESS in 4m 42s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/6f7adb2486574228a9bc7b67377614cc : SUCCESS in 15m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0fdd5e6061304fa19d17334ae386d106 : SUCCESS in 14m 36s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d37e5cd06e6841848540dd11dafa0845 : FAILURE in 32m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/95bb95bed3954512b1aedad9a1bd214e : SUCCESS in 52m 39s\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/fd4bf73e84594e58ace6cd5c44ef0bda : FAILURE in 4m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/72d9577648494599b43bf7b92d8ac2af : SUCCESS in 18m 45s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40da361d0a8c41fa8a3f894901487733 : FAILURE in 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6f0a7798ed4025ade72bea0dd6fb246a937779c6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-28 02:18:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70247c8738929b2c6fbfaa375566e9949dbb6c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 03:26:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/7273b96084eb41a48babaf0d470795e5 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/611291989e084deeaa49c5b42cb7564e : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/c9216430f6bc4b1e978f16011029cb9d : SUCCESS in 4m 41s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/b9cf9dd7698b485b949dc2dd4819e476 : SUCCESS in 22m 28s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b147256229414975b9c49caa0d91e8fa : SUCCESS in 20m 10s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb8ac4623a174ce3a824e26cefbc5921 : FAILURE in 33m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3224cd61c7204fb7b88eb635087e847f : SUCCESS in 1h 03m 12s\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/130cf06835484ee6af2063fee8da2325 : FAILURE in 4m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/72c191dd47ee4682a68001e69e661b47 : SUCCESS in 17m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29e754fe3a46428ba2f97beca00aa8ce : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"27680011ad06a674972fa1cb5fd6f80a4e324de8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-02 15:45:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15349b88531fd68bb1bea2123147f3e4e1ce08ec","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-06-11 09:06:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35749da16dc97ddb37db0a413fa3a6e8c99daae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 09:10:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"048c5d8918348ac0cb5c85cb5f31d3ee430e7a0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 10:10:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/2eacd0d5061a4f80bd14bdeaa8e20e19 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/f67a301f4b4949d1a2a401d67c3deff1 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/a2f01c5c805d488a8a0d4ea1681f30c1 : SUCCESS in 4m 44s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/0458a3963ccc4a938621433a9214fc17 : SUCCESS in 25m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/649e9f4319ad4b28817f9d641279eb3e : SUCCESS in 22m 22s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/58370eb63c384b3db1def669a7649a86 : SUCCESS in 56m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/974cdcbb2f1943d18ef5a733ab7a3d58 : SUCCESS in 19m 29s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ed80e46b31f4f4fa03f587701c9c984 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f19ae6aed96c12667ed6fce479ab320b4aa9c28e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 10:10:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41c74eeed59cabf34b5c6ad68b7609b2ddbc6979","revisions":{"41c74eeed59cabf34b5c6ad68b7609b2ddbc6979":{"kind":"REWORK","_number":1,"created":"2021-05-28 01:12:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/793520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/20/793520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/20/793520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/20/793520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/20/793520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/20/793520/1"}}},"commit":{"parents":[{"commit":"e887081f1da807d576c390fb663771e2a04f83f4","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e887081f1da807d576c390fb663771e2a04f83f4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-28 01:12:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-28 01:12:38.000000000","tz":540},"subject":"Rocky only: Remove TripleO Job","message":"Rocky only: Remove TripleO Job\n\n... because TripleO has EOLed its Rocky.\n\nChange-Id: I4ff93cf4a1f225f88750fd37a392d36cb4dce7bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/41c74eeed59cabf34b5c6ad68b7609b2ddbc6979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/41c74eeed59cabf34b5c6ad68b7609b2ddbc6979"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
