)]}'
{"id":"openstack%2Fpuppet-heat~798870","triplet_id":"openstack%2Fpuppet-heat~stable%2Fvictoria~I2c3a5d3c005963fe490310f6b81619cc85b28cc9","project":"openstack/puppet-heat","branch":"stable/victoria","topic":"at-params-stable/victoria","hashtags":[],"change_id":"I2c3a5d3c005963fe490310f6b81619cc85b28cc9","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2021-06-30 11:33:40.000000000","updated":"2021-07-01 03:54:06.000000000","submitted":"2021-07-01 03:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798870-at-params-stable/victoria","cherry_pick_of_change":758557,"cherry_pick_of_patch_set":2,"meta_rev_id":"350c5875fb455e1d23c2751d18d007fb7e71c107","_number":798870,"virtual_id_number":798870,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 03:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-07-01 00:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-30 14:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-07-01 00:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2021-06-30 14:45:56.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 11:33:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"state":"REVIEWER"},{"updated":"2021-06-30 14:23:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-30 14:45:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-01 00:42:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"282966eca9d98f112d1a8d84950dfe5099910c39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 11:33:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d2978f3d4d00b9a2746f54b22e23eca67eccea91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:23:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de9373f66d704bc183034ff25c87f213 : SUCCESS in 6m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bba3449ecb64c118c74ecc0f259d8eb : SUCCESS in 6m 30s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf7018d01a7040fe92538fe1afaa9512 : SUCCESS in 15m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ef3ed8e1d1f4405b7d53ba96afdf41f : SUCCESS in 12m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e2bb1c7a9bc469ea7effaa22de6efe3 : SUCCESS in 13m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/997fe480ff1741af845c3e5a7639f117 : SUCCESS in 49m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3f20640cd5284bc59588bfd6a692d8be : SUCCESS in 1h 21m 40s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e524f99ccf24612b7e5e158936ebcc4 : SUCCESS in 1h 22m 23s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/edf9252869cb45bcbf0333ef76dac468 : SUCCESS in 24m 34s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/68350650532244efb3e8cef5db1dd110 : SUCCESS in 28m 13s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/233a4d3acf284ebe9bbcac3a307841cf : SUCCESS in 20m 32s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2108676775f845c5aa4b7928e66bd922 : SUCCESS in 2h 44m 40s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5bbe8128e9814e1abb4365a9a599eedd : SUCCESS in 1h 21m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf8c111274bf403d884a433b0caec33c : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e34241b8d1268ca2e92e2bb2a1bdc0b1c25d781d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-30 14:45:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c862950c50b01728ccc2f740ec3a4264a68fdd3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-30 14:45:56.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"cc71f479967d1442cbad5d5f8192c4a62f4d34f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-01 00:42:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11c9fa3769a696212cc1e67ed12cd7105508cd53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 00:42:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c74f919623426d0ebc5a3da0df7205bd3d1f9c17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 03:52:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18b960a748e04e008005b0ab335ae997 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfea4621df5f43858fc44cb9a012a947 : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/c722cc782b5240f38a3aef8d7d778969 : SUCCESS in 13m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20d0629bc741458f84166c13b2fcbc88 : SUCCESS in 12m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/27dcd2bd29734f7fb5a8fdfcde81d84e : SUCCESS in 1h 22m 05s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e336796f3c2e4d66a1ce1b380cd3e4f4 : SUCCESS in 19m 20s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/79df1c5cf6f4447b82af9fdc8bc654a6 : SUCCESS in 25m 57s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cefd1d59b4ee4b898d1f16b5921fb6a3 : SUCCESS in 17m 50s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f7f55aaf44e24d198677be29760ae70e : SUCCESS in 3h 05m 18s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/dfa7eb49ec0c4409bba59b0afe15718b : SUCCESS in 1h 21m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69f325761ae7451eae865aca539c3ba3 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e18723fe7ecc04afd2c8ebc347b4781c85143bad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 03:52:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"350c5875fb455e1d23c2751d18d007fb7e71c107","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 03:54:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c76ea6c82a44598c8c7c75838745a0 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99c18e2e548fb05401d145b17f11fc8f68edbc2a","revisions":{"99c18e2e548fb05401d145b17f11fc8f68edbc2a":{"kind":"REWORK","_number":1,"created":"2021-06-30 11:33:40.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/70/798870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/70/798870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/70/798870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/70/798870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/70/798870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/70/798870/1"}}},"commit":{"parents":[{"commit":"b37eb48da361c2f881d7760f29c5a815e9ed1b1d","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b37eb48da361c2f881d7760f29c5a815e9ed1b1d"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2020-10-16 12:35:45.000000000","tz":600},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-06-30 11:33:40.000000000","tz":0},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I2c3a5d3c005963fe490310f6b81619cc85b28cc9\n(cherry picked from commit 261f4b1cf6afcd03f829304066494ab04d206221)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/99c18e2e548fb05401d145b17f11fc8f68edbc2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/99c18e2e548fb05401d145b17f11fc8f68edbc2a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
