)]}'
{"id":"openstack%2Fpuppet-heat~806152","triplet_id":"openstack%2Fpuppet-heat~master~I293c28cc9e7decc2149b44b8f9154f088ebf09db","project":"openstack/puppet-heat","branch":"master","topic":"authtoken-ut","hashtags":[],"change_id":"I293c28cc9e7decc2149b44b8f9154f088ebf09db","subject":"Do not test authtoken parameters directly","status":"MERGED","created":"2021-08-26 12:01:40.000000000","updated":"2021-09-03 07:57:35.000000000","submitted":"2021-09-03 07:57:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806152-authtoken-ut","meta_rev_id":"eb32f23110921f9b3d5491730c14e0042f804aab","_number":806152,"virtual_id_number":806152,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 07:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-03 06:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-09-02 15:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-09-03 06:50:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 12:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 13:46:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-09-02 15:55:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-03 06:50:10.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"4d29b715ba7e8ab2deaa0aa40b4d22610dc9e6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:01:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c5365b71682eef3dda3157ec7377e76112d60fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:09:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f71f97263c5002b41d0ba6a8d279e31fd8a2289c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:17:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bde94e68f5ddb9197efd67eb077aca5cef9bbcdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:38:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dd6079f60364e1da2121b12dac35327 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bca1246bb844473b839e08323327fce5 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea6a695889ab4486940495028d871e7c : SUCCESS in 6m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4076cbd2a3794fcbb51ac4613d38a11f : SUCCESS in 13m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a5bfc5892e74c1981b77d9f6384ed54 : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eff572cf6d8149aa88a50b5f93306cb5 : SUCCESS in 10m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a3456bd2e4d472bb7ee31fe724cf314 : SUCCESS in 12m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/049be7a0562d4f2b87d82222ad02068a : SUCCESS in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"62b98f60db15e20e68aa525ce70e630761b0e17b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 12:41:13.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"37c329420893d0fc5127fc4b4af028eb8485d49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 13:18:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aa03043b045f0979ed09ece1f9f4c4dd7c231d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 13:39:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c78828a1d524fcc808a1eff6aa54cfa : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d096e558ddd043e5988e14b60943b675 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25616951041b49039dd02603825b87e9 : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d433d1a353641dabc32815f6fda1609 : SUCCESS in 14m 06s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/083e276df06d47259dac94800a03409f : SUCCESS in 13m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c7728d020954a3fb6efda33cbd5ec59 : SUCCESS in 12m 41s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32ba73a8b79c4dad843f3824c7048dcf : SUCCESS in 11m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff3234ceb99241bfa389152867aa88ed : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0fb34429e376f130b4b6c4c13fe98b09f656048c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-26 13:46:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"364add6c1e3493c1241cb423d0e62262668a7010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-26 21:33:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"00c4b3e0e27c7432007fc2a2cfba1c1b9527f883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 21:50:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/400fb6c8c489499f8f7052c601bfd5aa : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42d5113dcb004f18af96610b2148c58e : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24d2baccf5f643deaf6d74a22970f265 : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/29952dcd9c1a4380a9da264fb84cc22f : SUCCESS in 12m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e76fce0b753b4f4899658ed760712e6b : SUCCESS in 14m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3edb0b273dec41fcb8734382857f7250 : SUCCESS in 11m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10cf698a73604272b04b6bb3c8dddc02 : SUCCESS in 10m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/772fe68bac914ad9831e49fdd08f2673 : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"92aa15c62e8531d8d1a362a93e7e96eedf5ee2d4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 15:55:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"86e99ace24139ca982c7fec4af301c0858b2dc3b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-03 06:50:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"339c197c567061436134962d53e2973aac2f5821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:51:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ca84b73b02ab7fb6a5a3f0cadf5d30cc3cbe3c5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:12:22.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4eb9c0e31aa406ba632c2ffadb31aa9 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d5bf7ced8324d219c32275d36f88c91 : FAILURE in 5m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c153ee2cac874d6cad14ca224d5cb8d4 : SUCCESS in 4m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ad15a75d7de46bda0707db52f4f7c7e : SUCCESS in 15m 12s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1dc2e6ad14214143b2b7bbdad0345f40 : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99537a7eaea14c4baa4b6a846f3e180f : SUCCESS in 12m 22s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26fc4a3665824127b2f4f682c1f5bcad : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":5},{"id":"bde33426d76e9c915c6ac0922e1142dedba6f530","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-03 07:17:44.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6e2aa686f4161009b9ea3eccea9c5047dfe88bd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:37:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ffe1421f824aeeaded7f9b545bf605 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1538a6d0a8834c518f5fbf7a31b440a7 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0526c6dfc20447abe1b87657d06d6be : SUCCESS in 5m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f29dfdf4c4f4b06ad9584217e910f9d : SUCCESS in 13m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ca461722fa6484297814cba444d5388 : SUCCESS in 17m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6265365967964ef68c609cbb39b0f690 : SUCCESS in 10m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbfc912ccddd43e2beb3a608410286a7 : SUCCESS in 10m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c596985bf0234389a2d8ab67988b0c78 : SUCCESS in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1e7db2c2325847844e7f870eb631a6415c015201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:38:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0eacce2fca41e041204f6f666f12d41f5f44fd4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:57:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f2a42000dd042758d287dd094fa6ec6 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e978995cd514e1db556837fb281e6c2 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/954705a91d9e40189cc9de2bc96e4168 : SUCCESS in 4m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a0ac09c25394836adc7ba4cb4bd970d : SUCCESS in 13m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9ea9b17f01c4cc9b2183c070650d5d4 : SUCCESS in 13m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2810434ef564987989f6e2fb05de931 : SUCCESS in 11m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21ada89b7e5d455cae8ebf881e60bb8d : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":5},{"id":"eb32f23110921f9b3d5491730c14e0042f804aab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 07:57:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"849af0e11f36f1cd6cbd0bda1ac01e600c84e0d1","revisions":{"c89f565f409be3622e5df982bb8ad801d4504fde":{"kind":"REWORK","_number":1,"created":"2021-08-26 12:01:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/806152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/52/806152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/1"}}},"commit":{"parents":[{"commit":"7cc1cf4a25668e49144a942be9e87ee5fe894301","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7cc1cf4a25668e49144a942be9e87ee5fe894301"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:01:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:01:30.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I293c28cc9e7decc2149b44b8f9154f088ebf09db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c89f565f409be3622e5df982bb8ad801d4504fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c89f565f409be3622e5df982bb8ad801d4504fde"}]},"branch":"refs/heads/master"},"787972b7c58142da4f9e1f1884ef65cd4e63a68d":{"kind":"REWORK","_number":2,"created":"2021-08-26 12:09:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/806152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/52/806152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/2"}}},"commit":{"parents":[{"commit":"7cc1cf4a25668e49144a942be9e87ee5fe894301","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7cc1cf4a25668e49144a942be9e87ee5fe894301"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:01:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:08:55.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I293c28cc9e7decc2149b44b8f9154f088ebf09db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/787972b7c58142da4f9e1f1884ef65cd4e63a68d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/787972b7c58142da4f9e1f1884ef65cd4e63a68d"}]},"branch":"refs/heads/master"},"ff4ae80185106033388087ac80a271433075a370":{"kind":"REWORK","_number":3,"created":"2021-08-26 12:17:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/806152/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/52/806152/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/3"}}},"commit":{"parents":[{"commit":"7cc1cf4a25668e49144a942be9e87ee5fe894301","subject":"Support Puppet 7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7cc1cf4a25668e49144a942be9e87ee5fe894301"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:01:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:16:56.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I293c28cc9e7decc2149b44b8f9154f088ebf09db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ff4ae80185106033388087ac80a271433075a370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ff4ae80185106033388087ac80a271433075a370"}]},"branch":"refs/heads/master"},"a191d39d848efebdaea42579443083b470183c71":{"kind":"REWORK","_number":4,"created":"2021-08-26 13:18:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/806152/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/52/806152/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/4"}}},"commit":{"parents":[{"commit":"79ee6b5c8fbd96fe98a6ab260b81f359c0febdb8","subject":"Use a \u0027params\u0027 hash for authtoken parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/79ee6b5c8fbd96fe98a6ab260b81f359c0febdb8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:01:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 13:18:24.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I293c28cc9e7decc2149b44b8f9154f088ebf09db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/a191d39d848efebdaea42579443083b470183c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/a191d39d848efebdaea42579443083b470183c71"}]},"branch":"refs/heads/master"},"849af0e11f36f1cd6cbd0bda1ac01e600c84e0d1":{"kind":"REWORK","_number":5,"created":"2021-08-26 21:33:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/806152/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/52/806152/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/52/806152/5"}}},"commit":{"parents":[{"commit":"79ee6b5c8fbd96fe98a6ab260b81f359c0febdb8","subject":"Use a \u0027params\u0027 hash for authtoken parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/79ee6b5c8fbd96fe98a6ab260b81f359c0febdb8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 12:01:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-26 21:33:35.000000000","tz":540},"subject":"Do not test authtoken parameters directly","message":"Do not test authtoken parameters directly\n\nThe authtoken parameters are not managed directly but managed by\nthe keystone::resource::authtoken class. Thus we should avoid testing\nparameters directly otherwise any change in the resource type can\ncause test failures.\n\nChange-Id: I293c28cc9e7decc2149b44b8f9154f088ebf09db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/849af0e11f36f1cd6cbd0bda1ac01e600c84e0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/849af0e11f36f1cd6cbd0bda1ac01e600c84e0d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
