)]}'
{"id":"openstack%2Fpuppet-heat~813033","triplet_id":"openstack%2Fpuppet-heat~stable%2Fxena~I5100dc896ffeaba2a1bc85079dd009521de4cde3","project":"openstack/puppet-heat","branch":"stable/xena","topic":"tripleo-xena","hashtags":["puppet-tripleo-job"],"change_id":"I5100dc896ffeaba2a1bc85079dd009521de4cde3","subject":"Remove TripleO job","status":"MERGED","created":"2021-10-07 14:22:48.000000000","updated":"2021-10-09 08:25:37.000000000","submitted":"2021-10-09 08:25:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813033-tripleo-xena","meta_rev_id":"903520e559d2975af6d74b7d425c9c3d385898ab","_number":813033,"virtual_id_number":813033,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-09 08:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-10-09 01:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-10-07 21:47:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-10-08 14:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-10-09 01:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-10-08 14:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 16:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 21:47:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-08 14:12:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-10-09 01:12:53.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"530a55355be0f2d00b8ecc5f0d9acca9cfe5014f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 14:22:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e59d6d85343c80dc8aa1d84953b66909b96f8fb","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-07 14:29:30.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"86d8782b36bbfba58e57a87dbfb3fe1a325a651f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 16:25:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7f80f762de543d289954c0359ddcdf9 : SUCCESS in 6m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11ef621cae9d422d8e55799455a356ba : SUCCESS in 4m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43294d433a6843f3a584fdc091d39a37 : SUCCESS in 5m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/136f27df2cde4c1392944c5076e6c594 : SUCCESS in 15m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/63957251b4ed47aaa2d4ec64436381c8 : SUCCESS in 14m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41348514c31345569a46ea99c485eadd : SUCCESS in 13m 51s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eb9859076fe403caf18c1c20033c7af : SUCCESS in 10m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e30d74b048e4b0fa7c54f907a6a83de : SUCCESS in 12m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73500fa1e09548ac968a98e5006b10c5 : FAILURE in 14m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f59a3ebc7954299bbadb6be1b20e8e4 : SUCCESS in 1h 42m 41s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1446f238faa34b48b96fb2f29a022fd2 : SUCCESS in 18m 24s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b56e0f6595048a398c6346f1fe70c6f : SUCCESS in 25m 56s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c425282912e4fe19a0b7de44c0f6711 : SUCCESS in 21m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3137baec539240abbf631bda93af9bb4 : SUCCESS in 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"196966b0579324561b970ed11bc502b758bf3de5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-07 21:47:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ed1909e23aac823735823fda647d33b26bd21c2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-10-08 14:12:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c9ddab0372c0252d65a9f16166d2dcc594465f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 14:17:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ac4e275d5d7e13392bf5cf735f56fe9e3e28b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 15:27:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fe8361d69f34255a50ec2b4108517d5 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4d5a725748d4c279473c92a03612175 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d072d8bd1ad43029a06158815c8b4a2 : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1444f10152854ba491b9a108ac36d281 : SUCCESS in 14m 03s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d0e8d5abd3f4e3fb4714305c578b093 : SUCCESS in 13m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cf58f8574a747f7b3c864ca83fb56a3 : SUCCESS in 9m 54s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a155777b31854f99b596221b381bb321 : SUCCESS in 11m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/988583ea706e4a6e8eddf71ddafb0c78 : FAILURE in 1h 08m 23s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/72daed4aa6c04bfc9425ac76db360561 : SUCCESS in 24m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cba884d0cae45a09da7b20b931fe66a : FAILURE in 19m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b5576a5c62751a8c1ee128776a4950da469a6b1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-10-09 01:12:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nreverify","accounts_in_message":[],"_revision_number":1},{"id":"539761dd8f26c8e42a9763fcdc1da7d190815cd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 02:49:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fa53a7952354e88bf26bfde53be40ad : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d659718e17424a4680457bb76e1ef712 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ded291fe2db940879ecbca370cd47570 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/53323534a2044cfdb05557e8678ccf43 : SUCCESS in 12m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcbbfc802b604bc99804129e68a91830 : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3754c85f6b804ab1833a1d3063f9fa4e : SUCCESS in 9m 57s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce88af15b3e7421c82a2e30380a82399 : SUCCESS in 10m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d6039f8515946ed99391c7ad20e5375 : SUCCESS in 9m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b3f2b0cf07c4df4951c61e69cbbf111 : FAILURE in 7m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a68b42618dc47949fe6fbe1603df521 : SUCCESS in 1h 31m 16s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf121e9e0d584b29a80829ea6c788791 : FAILURE in 16m 28s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a014cbb8cef44fc2bc7a1895c0082f51 : SUCCESS in 21m 25s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa206fa03d864004af62b3135b1e2e4e : SUCCESS in 17m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2fcc7f29ae54ed588acd252ee25d125 : SUCCESS in 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a297385d8a640c2c86f5ee941b0b1a517d543b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 02:50:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79544da9eae8833d46ed199a5f77cab471a4229d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 04:05:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eea498f3f2e74bb9aa50e5409eb6286b : SUCCESS in 6m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a406a4c5e2dc426eb3d07b0c6df87bf4 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f89bd3376524efeb8ef3755c21e32b3 : SUCCESS in 5m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/332c4d20776745a888571c6cdf30e588 : SUCCESS in 13m 39s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d86c9bfe6bf47df969f2be68a4f9137 : SUCCESS in 14m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b90206c6bc674ffd969c10d2daadf361 : SUCCESS in 11m 05s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5148f5a6fae417096f6f5f57a469226 : SUCCESS in 11m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c35e5dcf0344cf8b96a5b10bd60ecf6 : SUCCESS in 1h 13m 19s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a66f54277903476a9dd9ff5ed5f86248 : FAILURE in 35m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9931f168f1b744a48de0bac4c6c56c20 : FAILURE in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"353f3d9f3eb5f7f9beb91c3e35159880e18c5bac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-09 05:54:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c8852bf5e7a224086471f526f12854fdb83c6318","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-10-09 05:57:22.000000000","message":"Patch Set 1:\n\nreverify","accounts_in_message":[],"_revision_number":1},{"id":"b5d3386ec7108273361bf7d3dbfb921cc8c40288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 07:13:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ca539e52ed94c64a22ceb0895ede4a1 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2acfcaadc254de7a6aa97e6ea8db2be : SUCCESS in 4m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d3fda48ca14a95a56ded6461a0dfbb : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5865fc0696854c82b2c234f15235e54f : SUCCESS in 12m 29s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d9cc58c13564fde901f4f18d633637f : SUCCESS in 13m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1f93041475042e18bb64b043e14a3a4 : SUCCESS in 10m 11s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5610f30983e41c5b955ce4e2d303ec5 : SUCCESS in 10m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2fbf444b48d438b86b577c1e777ec9d : SUCCESS in 8m 01s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8247cbd605734240b736e63cc6c942ae : FAILURE in 10m 02s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/387ad392460441458ff9ff5dc2272477 : SUCCESS in 1h 16m 36s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8928da0baaeb4c0f8a36585b37296c4e : SUCCESS in 19m 28s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5ff35a4a47b41778b479e859cc4fff5 : SUCCESS in 21m 21s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dafb2bb959b4474b5b5d9169d1007d5 : SUCCESS in 18m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d4ab511d3d94707b7ef1b5c795a25ff : FAILURE in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"108106119be66ae3e3c94a86f9a36276371231d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 07:14:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e3288cdabc042966b7d446853ad4a9491a0c80a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 08:25:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8828899e47de4b7f99aebb4cd1e45187 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/740aee9ebd3647d79e16514e270a9bc8 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee289528b14b43ffa6478ca3f6472fb5 : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c50267f754c4329ad8497d01228a8e6 : SUCCESS in 14m 31s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1d9fb69529d42cf9b9c64f47e466d58 : SUCCESS in 13m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5498fde728241ce8bdd91ff974fbe7f : SUCCESS in 9m 40s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f7f5a6567d944bca8f722e900740761 : SUCCESS in 11m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/26a6af649da54134a1a6d2b290d9716e : SUCCESS in 1h 09m 58s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a0f0fcfa53d4123b33079007f96dc09 : SUCCESS in 22m 28s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c9364123e448d58dc116d8abb82eb2 : FAILURE in 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"903520e559d2975af6d74b7d425c9c3d385898ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 08:25:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"118f130d573e3edee371a209faee06895b5147e8","revisions":{"118f130d573e3edee371a209faee06895b5147e8":{"kind":"REWORK","_number":1,"created":"2021-10-07 14:22:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/813033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/33/813033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/813033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/813033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/813033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/33/813033/1"}}},"commit":{"parents":[{"commit":"3ed90d84503211489313abd70707c11c405a940a","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/3ed90d84503211489313abd70707c11c405a940a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 14:22:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-07 14:22:19.000000000","tz":540},"subject":"Remove TripleO job","message":"Remove TripleO job\n\n... because TripleO has switched to the independent release model and\nits Xena release is not being created.\n\nChange-Id: I5100dc896ffeaba2a1bc85079dd009521de4cde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/118f130d573e3edee371a209faee06895b5147e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/118f130d573e3edee371a209faee06895b5147e8"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
