)]}'
{"id":"openstack%2Fpuppet-heat~820157","triplet_id":"openstack%2Fpuppet-heat~master~I523e4a985e28599367962f595ceaf66ca511276f","project":"openstack/puppet-heat","branch":"master","topic":"puppetlabs-apache","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-08 13:14:35.000000000","reason":"Change was abandoned"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2021-12-07 16:33:20.000000000","reason":"Alex Schultz replied on the change"}},"hashtags":[],"change_id":"I523e4a985e28599367962f595ceaf66ca511276f","subject":"Add missing puppetlabs-apache to dependencies","status":"ABANDONED","created":"2021-12-02 12:19:17.000000000","updated":"2021-12-08 13:14:35.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"89c39cd97d71cdd6ac5f1f675cac6b0aa87b06da","_number":820157,"virtual_id_number":820157,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-02 12:24:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":-1,"date":"2021-12-07 15:52:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 12:24:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 15:52:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"057ee1cb4ce468bb986c64f4b4102e080e72e786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-02 12:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7b78520377ae488d6feea3938023e844a972ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 12:24:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/74e33197376549c5a9cf0a6a921b42ef : SUCCESS in 2m 56s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a82255e1fa6c4656961b5212cd5d86ea : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4fd9b3a8d7c76cdcc2989bd510c79d1faf8e1618","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:52:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1be1d73d126de55f972a2017848014b4ec2bac7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 16:27:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87aeaee458e5387a50d7b63fb7ff5412d5ea90d3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 16:33:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89c39cd97d71cdd6ac5f1f675cac6b0aa87b06da","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-08 13:14:35.000000000","message":"Abandoned\n\nIt turned out the remaining direct references are just redundant. I\u0027ll propose patches to drop them so that we can remove puppetlabs-apache from dependencies of individual modules.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbce87f79c7429830ac0e08a5777a29a6f3cfc18","revisions":{"fbce87f79c7429830ac0e08a5777a29a6f3cfc18":{"kind":"REWORK","_number":1,"created":"2021-12-02 12:19:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/820157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/57/820157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/820157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/820157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/57/820157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/57/820157/1"}}},"commit":{"parents":[{"commit":"b3685e6e136520a82076f1d822b2311b191f4551","subject":"Use consistent spelling, OpenStack, instead of Openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b3685e6e136520a82076f1d822b2311b191f4551"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-02 12:18:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-02 12:19:04.000000000","tz":540},"subject":"Add missing puppetlabs-apache to dependencies","message":"Add missing puppetlabs-apache to dependencies\n\nCurrent puppet-heat depends on puppetlabs-apache to support running\nheat-api and heat-api-cfn using httpd+mod_wsgi.\n\nChange-Id: I523e4a985e28599367962f595ceaf66ca511276f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/fbce87f79c7429830ac0e08a5777a29a6f3cfc18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/fbce87f79c7429830ac0e08a5777a29a6f3cfc18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
