)]}'
{"id":"openstack%2Fpuppet-heat~826850","triplet_id":"openstack%2Fpuppet-heat~stable%2Fwallaby~Ibbda48c1935e3a26e6f4a2d39ed603383a5adc6c","project":"openstack/puppet-heat","branch":"stable/wallaby","topic":"wallaby-c9","hashtags":[],"change_id":"Ibbda48c1935e3a26e6f4a2d39ed603383a5adc6c","subject":"Prepare a new Wallaby release","status":"ABANDONED","created":"2022-01-28 08:50:00.000000000","updated":"2022-01-28 09:24:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1f1c44589a828265241f9d53a223671e83b237cf","_number":826850,"virtual_id_number":826850,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-28 08:59:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-28 08:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2bb069126e0149395b032f459e265ffdefe68df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-28 08:50:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9475cfcceaa204277450a0768c5906b031ad5d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 08:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/edc65df4d988475782f18e871369afb2 : SUCCESS in 3m 01s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc2bdec5365a4ba2b05ffcb055ff7a4a : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1f1c44589a828265241f9d53a223671e83b237cf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-28 09:24:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8350a421383e308677bd6bc98587772f416283be","revisions":{"8350a421383e308677bd6bc98587772f416283be":{"kind":"REWORK","_number":1,"created":"2022-01-28 08:50:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/826850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/50/826850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/50/826850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/50/826850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/50/826850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/50/826850/1"}}},"commit":{"parents":[{"commit":"62e45b088e31cb891c4d003df62601a0d955e3d0","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/62e45b088e31cb891c4d003df62601a0d955e3d0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-28 08:49:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-28 08:49:41.000000000","tz":540},"subject":"Prepare a new Wallaby release","message":"Prepare a new Wallaby release\n\nWe recently merged several fixes for CentOS9 support. Let\u0027s create\na new release so that these fixes are included in a released version\n\nChange-Id: Ibbda48c1935e3a26e6f4a2d39ed603383a5adc6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8350a421383e308677bd6bc98587772f416283be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8350a421383e308677bd6bc98587772f416283be"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
