)]}'
{"id":"openstack%2Fpuppet-heat~828304","triplet_id":"openstack%2Fpuppet-heat~master~I2d7619574f93958ceae15bafea02d77554fb5315","project":"openstack/puppet-heat","branch":"master","topic":"amqp_allow_insecure_clients","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-15 10:56:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d7619574f93958ceae15bafea02d77554fb5315","subject":"Remove deprecated amqp_allow_insecure_clients","status":"MERGED","created":"2022-02-08 13:07:56.000000000","updated":"2022-02-15 10:59:20.000000000","submitted":"2022-02-15 10:56:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828304-amqp_allow_insecure_clients","meta_rev_id":"f19a0d132804cd57523dd6cfa33103c52a56a6a4","_number":828304,"virtual_id_number":828304,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 10:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-14 17:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-15 08:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-15 08:57:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 15:52:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 17:48:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-15 08:57:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b6323a68ec7a2e89a539a4927ea7bfbf3d048c08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:07:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d3f4d8850a746313a0b34b04eb453d459573c9b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:21:34.000000000","message":"Topic set to amqp_allow_insecure_clients","accounts_in_message":[],"_revision_number":1},{"id":"665f0566519d1a1adc9fa1e45ba980b63ed65103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:52:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf0961663426499eb9a9c917ca4f500e : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6b60bfc1b384a309794721608b4ae2c : SUCCESS in 4m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a1ddf43a65b42ab8d5c367a0e9c5542 : SUCCESS in 5m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3ee2a278b6340f8b13c40b6a49f8051 : FAILURE in 13m 54s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/856bc124e7eb4ac1b0375313b3e5e3a3 : FAILURE in 13m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a026d4b2c6e0423c82e86edb70e42aa5 : FAILURE in 11m 16s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12fb6231a747401b9e7f27641fae1eb4 : FAILURE in 10m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/294c375081d04fecbfd1b8b7e52dfdaa : FAILURE in 10m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/317b833d8b594167891ed08886a2f13d : FAILURE in 33m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f99a6c3689e44cfa98121faf9f69d4bc : SUCCESS in 1h 21m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb602fd5640047088931019877045d73 : SUCCESS in 1h 00m 13s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8614034d0ad2469aad8b8f124d38f503 : SUCCESS in 17m 59s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b1aaf9d0af64b6196753d0ef4ff378a : SUCCESS in 22m 56s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dbe8ecaa6304f5cb06e20f938403140 : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ef179635dd74121bf14aca45e5f2d5b : SUCCESS in 17m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d9c0cd6283a44e0b1dba00cf5988404 : SUCCESS in 18m 05s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cb5e24ca8d9b47d3a22daed969215794 : SUCCESS in 2h 37m 03s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a58ab57ce6c84da594d8431781b1d23e : SUCCESS in 53m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54578d5716d34ef3aee6b426d26516f8 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b51a3983263e7127b6a746c7d255a452b80e93c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-09 01:00:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a3be8da5cedc0e2071f19cf7dd2c1984bc905aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 02:42:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da03e09edba9485f85d13533fab33f7f : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c035aa9d0684911a19f79222db14e8f : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/472b821c82eb4bbda0e7d3c406bd3a8d : SUCCESS in 2m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8c6e49f661e4e9faba00354ae75e7f3 : SUCCESS in 9m 49s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8a066615d8b4d8f85e04879ff1d83e7 : SUCCESS in 9m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e439618a0a9c4a66bae79899cac70f19 : SUCCESS in 7m 51s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a010490ba8ab45379f4db88ccd36c8a1 : SUCCESS in 7m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8089a8a282984e27a4aa4e17f34cc0bd : SUCCESS in 7m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2d0a0e67d2b41558f085f4995e7f6fe : FAILURE in 47m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7822e52d2273430b9d9b5352748c29f2 : SUCCESS in 1h 39m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6d7570ff0394a06935588ff78e91c91 : SUCCESS in 1h 09m 58s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73e90772e26a4d699003ae9de51de8e0 : FAILURE in 12m 23s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb269f5f9a1a4419b6f0dc2af468ee0a : SUCCESS in 24m 01s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5068466fae34c74bd19fd2d45a323bd : FAILURE in 9m 57s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e30719469994df2ab35e033a264ca7e : FAILURE in 12m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bf3391dc5dd4fe081486de442bb44e9 : SUCCESS in 14m 03s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/94b4d2399f954c9bb7d8282a596c8bac : FAILURE in 58m 11s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e4e7e7a5b43a4fe99e1a8ddbbc52edaa : SKIPPED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9165182e79d645d084e080723b532a1c : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f9bfdf2c56dbfdd0d567e094c28a5cb71e9a3d86","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-10 02:44:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a3e0a91322a10bcc4ea2818a13df46f3d09e284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 05:44:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57fc0f3686004772bf195d790ff2034d : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11e0bc337cf54ac7990113e576e31325 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0048d99567af40e6970028d50cde9432 : SUCCESS in 4m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7ae7d312e8f4ebebac83d9dadc0d6c3 : SUCCESS in 14m 26s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2c1d5c2199946bf9bad1b4998d7ed07 : SUCCESS in 17m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4ea5af8634a42bfac8ea91ad4811f00 : SUCCESS in 11m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b46f41ecae0144518ee0e94259463369 : SUCCESS in 13m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79eb54e6b37341009bded12c90319716 : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94971f4c788940bb837865d026a14b6f : FAILURE in 30m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd4f86bfaba649da807a39329763ad38 : SUCCESS in 1h 07m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02b926e89e8d43239f205fd25160955d : SUCCESS in 1h 01m 37s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f50f8664a0457b932ff77f48fb73a9 : SUCCESS in 19m 51s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9faac60fef934d478b7e5943a43cc000 : SUCCESS in 23m 54s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c5c354fd71a4479821806f07eab5d68 : FAILURE in 13m 14s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b4370f6e1a947b9b81af557278cf958 : FAILURE in 15m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec7991c40a164513ae8eb113351848e3 : FAILURE in 15m 02s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/668fca58da9a4bc287a6a07e9b3fc2e3 : SUCCESS in 1h 52m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cec0386da00b4d77a8fe8afa541397d9 : SUCCESS in 42m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bc6a6f5522b4f6d8cb38e7d568ded3b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2df799f038510af3b2bbf235ffa958528a7f5bad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-14 17:48:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e22baa039be0917a9e28b210d19d2c3cfda19a9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-15 08:57:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d3bf3b5418b956fe1195175824b0f58a21974547","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-15 08:57:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b05c1c59042670b511b974c4ff7526a8ce57ec43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:59:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03f560e188d18b74663838b8301f6bd7b350c867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:56:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a55acbd35e427d95cabb379cb85ca3 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ed13f7dc6b348938dc6ed7b3a3b72e7 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56a038d5765441849d1fe98515b5a76c : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2caf488cc68243b3bfcde7b02b26b12f : SUCCESS in 12m 36s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/edfb4407310944a18a35b0b915cead51 : SUCCESS in 14m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40bda9531d944560b223834fe8312fdb : SUCCESS in 11m 06s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ecb965e7fac4039a31edc10c3d7511a : SUCCESS in 11m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ead08190f4e94145ba65876cfaab821a : SUCCESS in 1h 44m 05s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a52ec0505024191a850ac8195ddade4 : SUCCESS in 22m 01s\n- tripleo-puppet-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/fe33f40aed9649948141418c1d48ca39 : SUCCESS in 1h 14m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/412d9ce3cbbc4d4d8389d76e1bd5d5c5 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"4dc16e4204bb705370f01e22fd0bad41c68736ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:56:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f19a0d132804cd57523dd6cfa33103c52a56a6a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:59:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30788c350d154736996fc8e073345728 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37d5b04d7841ac1420d098f412f114a823f4a893","revisions":{"8dc916998e3e22d51ba8394dee0af42b5c370811":{"kind":"REWORK","_number":1,"created":"2022-02-08 13:07:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/828304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/04/828304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/1"}}},"commit":{"parents":[{"commit":"96c1ca67b5a0e7b5d563351ed889b72475382a49","subject":"Remove some deprecated database parameters left","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/96c1ca67b5a0e7b5d563351ed889b72475382a49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:07:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:07:28.000000000","tz":540},"subject":"Remove deprecated amqp_allow_insecure_clients","message":"Remove deprecated amqp_allow_insecure_clients\n\n... because it was deprecated during Wallaby cycle.\n\nChange-Id: I2d7619574f93958ceae15bafea02d77554fb5315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8dc916998e3e22d51ba8394dee0af42b5c370811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8dc916998e3e22d51ba8394dee0af42b5c370811"}]},"branch":"refs/heads/master"},"37d5b04d7841ac1420d098f412f114a823f4a893":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-09 01:00:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/828304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/04/828304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/04/828304/2"}}},"commit":{"parents":[{"commit":"ff16bee10a982cea38d82e38965178e9dc94ce3f","subject":"Remove some deprecated database parameters left","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ff16bee10a982cea38d82e38965178e9dc94ce3f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:07:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-09 01:00:20.000000000","tz":540},"subject":"Remove deprecated amqp_allow_insecure_clients","message":"Remove deprecated amqp_allow_insecure_clients\n\n... because it was deprecated during Wallaby cycle.\n\nChange-Id: I2d7619574f93958ceae15bafea02d77554fb5315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/37d5b04d7841ac1420d098f412f114a823f4a893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/37d5b04d7841ac1420d098f412f114a823f4a893"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
