)]}'
{"id":"openstack%2Fpuppet-heat~832889","triplet_id":"openstack%2Fpuppet-heat~stable%2Fwallaby~I49413dac12d73d1e4315587a3dbe58c7ecb9c49b","project":"openstack/puppet-heat","branch":"stable/wallaby","topic":"c9_wallaby_gates","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-03-10 07:18:09.000000000","reason":"Marios Andreou replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-11 10:24:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I49413dac12d73d1e4315587a3dbe58c7ecb9c49b","subject":"Add centos-9 jobs for wallaby zuul layout","status":"MERGED","created":"2022-03-10 04:04:50.000000000","updated":"2022-03-11 10:24:41.000000000","submitted":"2022-03-11 10:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"832889-c9_wallaby_gates","meta_rev_id":"a38102d31299806e8430e4a8b6d2840b0c6cbcdb","_number":832889,"virtual_id_number":832889,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 10:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2022-03-10 07:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-03-10 23:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-03-10 23:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 04:22:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-03-10 06:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 23:43:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"80e6910dd3b45b2e91c9de95036353f036188588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-10 04:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b454ddbf6edad763d7fb06a19854f3bd2083663b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 06:04:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/594e1e2d64fc4db0a505b50e76039942 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7b4ec1a51154117ac9db2dfad6a4765 : SUCCESS in 5m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4271829871140f9898648208cced85b : SUCCESS in 13m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11c2aea9327d45b2a4339818bd277110 : SUCCESS in 11m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b7e5b20714d458caaa21c7af9e593e3 : SUCCESS in 11m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d34f6b5611314e80a83f97014bf4dd22 : FAILURE in 36m 16s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f8769aba60348c8a1ca34a64ef1c709 : SUCCESS in 1h 28m 04s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/031803dc3d0b4aa78be693250c4a8660 : SUCCESS in 22m 29s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a85e98848c34384b8c3f1724437d648 : SUCCESS in 20m 28s (non-voting)\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9baf2f9ee1464be09752b4445a180231 : SUCCESS in 1h 56m 12s\n- puppet-heat-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f64619e998274e6e9068991a3395be35 : FAILURE in 1h 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5feffd4c242c42e4e8ea3002e4ba452b76d6c08","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-10 07:18:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f01217a4613f2286c058e5d54da9b1b7b8cb7d2","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-10 07:18:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d7d64801e9a364d49547c9e1eb1f6c1f65faa2d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-10 23:43:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb4700d5801b48152dceb59adb5691556249713c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 08:12:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b6dd2426564f7c3fbe8bded2805e8a38b05a53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 10:24:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a24ae449bf3340389cdff679ed96497f : SUCCESS in 5m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c5c8f262bd64429b0da5ffe4ed8a103 : SUCCESS in 4m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9572aa64b2094217b5c1c53ebd5d34d4 : SUCCESS in 15m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbca29dab5b04bf785b80ebe3ba76b9c : SUCCESS in 10m 18s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6288e5da44684832b122e90b8221a05f : SUCCESS in 1h 20m 43s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/4deb9d841103400badc838583a7567a7 : SUCCESS in 26m 04s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f2e6eb240ae44feba357b8eb6acb6af : SUCCESS in 16m 30s (non-voting)\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/795dfd3fbd82415db1db2a8da2812730 : SUCCESS in 1h 59m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a38102d31299806e8430e4a8b6d2840b0c6cbcdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 10:24:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e4384da1140241c9ec0ddab0d7dded03e0df48d","revisions":{"7e4384da1140241c9ec0ddab0d7dded03e0df48d":{"kind":"REWORK","_number":1,"created":"2022-03-10 04:04:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/832889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/89/832889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/832889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/832889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/832889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/89/832889/1"}}},"commit":{"parents":[{"commit":"74009f0d425c5b7f4e6c37a88737e2cc489102c0","subject":"TripleO: Replace undercloud job by standalone job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/74009f0d425c5b7f4e6c37a88737e2cc489102c0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-10 04:04:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-10 04:04:36.000000000","tz":540},"subject":"Add centos-9 jobs for wallaby zuul layout","message":"Add centos-9 jobs for wallaby zuul layout\n\nAdds puppet-nova-tripleo-centos-9-standalone definition\nand into the check zuul layout. This is added as non voting\nas puppet openstack will not support c9 on wallaby.\nSee the topic branch [1] for related reviews and [2] for context.\n\n[1] https://review.opendev.org/q/topic:c9_wallaby_gates\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027403.html\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/831206\nCo-Authored-By: Marios Andreou \u003cmarios@redhat.com\u003e\nChange-Id: I49413dac12d73d1e4315587a3dbe58c7ecb9c49b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7e4384da1140241c9ec0ddab0d7dded03e0df48d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7e4384da1140241c9ec0ddab0d7dded03e0df48d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
